)]}'
{"id":"openstack%2Fglance~842595","triplet_id":"openstack%2Fglance~master~I0e2bca4603a34e39ffdbebaab2fcfad3a9da4241","project":"openstack/glance","branch":"master","topic":"drop-py3-6-7","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-23 09:52:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e2bca4603a34e39ffdbebaab2fcfad3a9da4241","subject":"Add openstack-tox-functional-py38-fips job","status":"MERGED","created":"2022-05-19 18:09:11.000000000","updated":"2022-05-23 09:57:30.000000000","submitted":"2022-05-23 09:52:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842595-drop-py3-6-7","meta_rev_id":"cc9fa4e0f0351af999371aec5ee9cbcb07fe3f31","_number":842595,"virtual_id_number":842595,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-23 09:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-05-20 02:54:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-05-23 06:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-20 06:21:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-05-23 06:00:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-05-20 02:54:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-19 19:17:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-20 02:54:52.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-05-20 06:21:20.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-05-23 06:00:10.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"351f27b31b0ccf67e90f35d80cb05b7509a2daf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-19 18:09:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60f767d0d4b848a0590b10eadf29a14c10aeb952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:17:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f7dce04b470424d988735387d15743f : SUCCESS in 1h 01m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/128c211859a948d6992a033be7d8986d : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43b933764ab74bd0afebc91bec071adf : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d021e4d5faa437fbf13f4f3fa2ce134 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61a2485edd014c3b8c4905baa4d503a8 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61881f63d97d4a2ab98baebbcacbf535 : SUCCESS in 7m 53s\n- openstack-tox-functional-py39-fips https://zuul.opendev.org/t/openstack/build/9f85f6ee87414ab2becf04be841e5c51 : RETRY_LIMIT in 4m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae0f532e6d594d5ab9e0fa253af54565 : SUCCESS in 9m 52s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/36885c276ef84f8e8bd911e0e83df50b : SUCCESS in 9m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d767e79a67284e7daeb75fb4bca05ae6 : SUCCESS in 44m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8d4eac36495229a491200635c8f52ce3a9ee6e06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-20 00:06:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"96d3160daf5736790f95f57ab80aeba34c562a89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 01:09:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/645c2f1dc2364e5ab75c7ec73cf3a21a : SUCCESS in 57m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8773bc4b63e24e41801baeeba7c1951c : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f8d9f3ccd584c419f6153febf3dc244 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83775b9cf2f44f47981c5eab1c5cfd07 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6055c231f32f47dfb541af8144ba9d24 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ffd331a9b9a4d988502f3ed471b6365 : SUCCESS in 6m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1d47973d5ed44d8b9afdb05a92b797b6 : SUCCESS in 11m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a4f718ae5a8d41548562cb3e1ad30217 : SUCCESS in 9m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b999e25dc3c84509b19a914609f237f7 : SUCCESS in 9m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d37bbad7b6414439a033778dbf6d717d : SUCCESS in 44m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3cd84da95132abd0fae721dff9c2e521048beda5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-20 02:54:52.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c7c0bdc359ffa7f28bc28018e066086f9790d260","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-05-20 06:21:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2a0ad69abcc6332f54c3251a0786d4f2d763f74d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-23 06:00:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2e84a1027a26c12d2e46dc3a71b6b7959d210d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 06:00:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"751c8626d8b830ca8728041f2e83fbda189e4a36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 06:48:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c46edad8619b4b85b3dd220152cad4c3 : SUCCESS in 46m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61dde248345e42f5a8c213f63cd1787c : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26985aed73d143528be8a695761ee114 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8499b3bd50634a839fc0efe5c496f931 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65136288a32048cd9ed2609db53cfc4c : SUCCESS in 6m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c124e19203e49f588ef67f589968048 : SUCCESS in 8m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f4de8b362d5d464b9671c6ab9e089355 : TIMED_OUT in 31m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/11ea59e3bd434e30ad0a97dd815c89ce : SUCCESS in 28m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b145ca7dcbab145a6a6e6ccb0170f806eb9cc049","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-23 06:51:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89f75d5b0f8d7763866ab3fbaf07777a7870bdcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 07:55:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71b3e1f4bfe74b7cbe52b0443d9e3c42 : SUCCESS in 1h 02m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a91b55e8cff4248a7b7aee458fcd2ce : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/523f1e2a2d0f43afb241d7983aae7cfb : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f8790bb6c3c4cec8901ecdd07f8c60a : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fde49393bc6045248b2d302ce851827f : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be59e1baac39451fa94620f2a1d75e5f : SUCCESS in 6m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0e1e7f27cdae4b6ab0feb26e3dba299c : SUCCESS in 12m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c675aaaa3db447cf9f5abec80b715403 : SUCCESS in 10m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/75f9b068de474053970a2c13cea99432 : SUCCESS in 9m 25s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/791fce3e86c64f28a8a849768edf37b6 : SUCCESS in 37m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a6d8c911bd4a8b04b05f7e960f80e8f03e11d152","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 07:55:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"be05eae2d9e9535e4432d65bdd45471f4ff6849f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 09:52:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d59097eaff9480aa80297df59051ee0 : SUCCESS in 41m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e69d2d7af25b4288bd0619c76d80f7db : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd3e18c325e449b4a279fe087e21b63e : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1d7fb61edb1446794d63c9bfa579f7e : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea40c32706a847bc931ba48b6007cee4 : SUCCESS in 5m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e2e5d29ab7744c87be831df0d14b0cb2 : SUCCESS in 7m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/683dca5d6097457996ba8b2e89eb2a8e : SUCCESS in 8m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/856b6b8aa3514f07881f497be25984b2 : SUCCESS in 31m 06s","accounts_in_message":[],"_revision_number":2},{"id":"3bee9e918d808be4dc13385ba1734f68f5825c21","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 09:52:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cc9fa4e0f0351af999371aec5ee9cbcb07fe3f31","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 09:57:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/303a57595dd94f0e9e22730dcfc8eb84 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b8e6bb0277231b51c10bf0570a647e0060ab8b03","revisions":{"448d9ea48787a1855a148032c9e4b87547fb3895":{"kind":"REWORK","_number":1,"created":"2022-05-19 18:09:11.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/842595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/842595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/842595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/842595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/842595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/842595/1"}}},"commit":{"parents":[{"commit":"d55d250d336edb1892f0f40cf8be6063be3315a4","subject":"Update python testing as per zed cycle teting runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d55d250d336edb1892f0f40cf8be6063be3315a4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-19 18:07:35.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-19 18:07:35.000000000","tz":-300},"subject":"Add openstack-tox-functional-py39-fips job","message":"Add openstack-tox-functional-py39-fips job\n\nopenstack-tox-functional-py36-fips is removed in\nhttps://review.opendev.org/c/openstack/glance/+/841350/\nand planned to be move to py39.\n\nThis commit migrate the fips job to py39 and c9s based.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/842593\nChange-Id: I0e2bca4603a34e39ffdbebaab2fcfad3a9da4241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/448d9ea48787a1855a148032c9e4b87547fb3895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/448d9ea48787a1855a148032c9e4b87547fb3895"}]},"branch":"refs/heads/master"},"b8e6bb0277231b51c10bf0570a647e0060ab8b03":{"kind":"REWORK","_number":2,"created":"2022-05-20 00:06:02.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/842595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/842595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/842595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/842595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/842595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/842595/2"}}},"commit":{"parents":[{"commit":"d55d250d336edb1892f0f40cf8be6063be3315a4","subject":"Update python testing as per zed cycle teting runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d55d250d336edb1892f0f40cf8be6063be3315a4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-19 18:07:35.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-05-20 00:06:02.000000000","tz":0},"subject":"Add openstack-tox-functional-py38-fips job","message":"Add openstack-tox-functional-py38-fips job\n\nopenstack-tox-functional-py36-fips is removed in\nhttps://review.opendev.org/c/openstack/glance/+/841350/\nwith the plan to be move to py38.\n\nThis commit migrate the fips job to py38.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/841368\nChange-Id: I0e2bca4603a34e39ffdbebaab2fcfad3a9da4241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8e6bb0277231b51c10bf0570a647e0060ab8b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8e6bb0277231b51c10bf0570a647e0060ab8b03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
