)]}'
{"id":"openstack%2Fglance~843208","triplet_id":"openstack%2Fglance~stable%2Fyoga~I0cf35ed7e21497826cd581e59aa58774f030b9f6","project":"openstack/glance","branch":"stable/yoga","topic":"bug/1971521-stable/yoga","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-06-03 11:23:30.000000000","reason":"Brian Rosmaita replied on the change"}},"hashtags":[],"change_id":"I0cf35ed7e21497826cd581e59aa58774f030b9f6","subject":"Bump Image API version to 2.16","status":"MERGED","created":"2022-05-25 15:55:35.000000000","updated":"2022-06-03 12:52:39.000000000","submitted":"2022-06-03 12:51:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843208-bug/1971521-stable/yoga","cherry_pick_of_change":843028,"cherry_pick_of_patch_set":2,"meta_rev_id":"5f69caec0dc19abb02e748a20645513906a34a6b","_number":843208,"virtual_id_number":843208,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-03 12:51:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2022-06-02 11:57:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-05-25 20:24:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-26 02:31:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2022-06-03 11:23:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-25 17:33:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-25 20:24:54.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-05-26 02:31:38.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-05-26 06:13:43.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-06-02 11:57:25.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"5ea38e9fae210b0f5f5ef216deea5ada481005b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-25 15:55:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3de331fd839e64ba64e3da0e14df59f3c1fed23a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-25 15:57:58.000000000","message":"Topic v2.16-stable/yoga removed","accounts_in_message":[],"_revision_number":1},{"id":"ec98ce58e68818b9c0391f4e985ce3beb0d48be8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-25 15:58:01.000000000","message":"Topic set to bug/1971521-stable/yoga","accounts_in_message":[],"_revision_number":1},{"id":"3e8d0a1f28d62a3bc7700631dcb64419e66934d1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-25 15:58:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03a0ea38bc8cc5a70814b18a996566752b5fd521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 17:33:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e63c6e2a0f4b4832a77a5013348372cb : SUCCESS in 58m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/534aef9ff4c1479e947817f2b61a9ee7 : SUCCESS in 57m 35s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb472876be424b2c80f2926aa8ebbf0a : SUCCESS in 1h 16m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2d9faef0f7547d88d79880988adbc0d : SUCCESS in 55m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd269d4d8e61403aacb29ec14d737c6e : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd0ab84c37b84ba9ae82630c4a8ed83c : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28a929217293496e826d4aa4ab74d192 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e371d74a4fd4f5fa7d5a7df2229af63 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39d64077cb7d4adc926857b67f5614f2 : SUCCESS in 6m 20s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/c3556fb044374358b71d8e888d11dd74 : SUCCESS in 12m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4e41c5a4843245ec992aa072b62bff36 : SUCCESS in 10m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b4cb8fb6981846feabaa9fa41ea38dc5 : SUCCESS in 12m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1c4d2c14cd6f4ee3a7e7721e42b83d8e : SUCCESS in 5m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/edf1f52a69a048d29849818a5bfa52dd : SUCCESS in 1h 28m 36s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9d7c925275a649c2a7d3bdae6df4ca49 : SUCCESS in 1h 24m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b43d2e5412f54f30b263654ee47b4ae6 : SUCCESS in 1h 27m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5764d815295f4040b8de8d109adf92b3 : FAILURE in 1h 22m 16s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/9e6a917956da4254b623416a18ae198d : FAILURE in 31m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/467e7548307c464088121e8b91e496a6 : SUCCESS in 47m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3d8c4848534940f68ef2560ab439be12 : POST_FAILURE in 1h 19m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/066c519024ad4f77a0262f5ececeeb4f : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":1},{"id":"78b8443b01ac5efe760816eb70848df4640f5830","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-25 19:38:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"af0b17ba5410a34468dd3bd784319e4450a81244","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-25 19:44:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78d65c48c0a047b5e70dcb07cc14532f705284be","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-25 20:13:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"933352f6fa1d5ab4b4cc74f98e8f85a6830a3f01","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-25 20:24:54.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for the backport!","accounts_in_message":[],"_revision_number":2},{"id":"950fc9adb554ac6471e768af711b9763c2720cce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 21:06:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/dded1a5a56d044bb924fbef6b906b300 : SUCCESS in 50m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/03ccf8bfb77c43df83b702701ccc8123 : SUCCESS in 56m 15s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/94aef05ee2614498be0fb155292a2cc1 : SUCCESS in 1h 10m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6f5353728c2445e85c2799f93ea0e5f : SUCCESS in 50m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3cde354d3fa46acafc52db400c7cb5c : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc32d58b886244b9b4e6c9c5e59cef67 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be587c3903fc427abfaa2ebf7e9d5086 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d7d3cbfbe3d429994a92387b0615fdd : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c63a59bca07343158ba18d4777929491 : SUCCESS in 6m 19s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/86a94ed035bc4d9dbf547727b5f3ea42 : SUCCESS in 10m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c3893ef7e3bd4b72a64c0f19eaac3894 : SUCCESS in 10m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6b1fcb73a5aa46b68bbfa3d61d333e1d : SUCCESS in 9m 33s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e0c12c00157a4bfcaaaa1dff329a5555 : SUCCESS in 5m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cc5563800033466e8af4795b825dc513 : SUCCESS in 1h 13m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/915ab84550294dcaa5fdf28dad4456a7 : SUCCESS in 1h 06m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e205b7c429494a5a8e1a215558970430 : SUCCESS in 1h 14m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/da364ccdc20d48cab023c429c8c9126a : FAILURE in 1h 19m 24s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/0418e880f9dc44739f30acd09caed576 : FAILURE in 32m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81dca75c2ecb4c3380157a807a802732 : FAILURE in 18m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/81d53c882e214959b22971ff77ad5f01 : SUCCESS in 1h 21m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/906f8a29363344419d9037a5563bb30f : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":2},{"id":"1b6092d891243fb4a564858ffa32397fa9902d67","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-25 21:58:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c69fc5bbd0f5aac889b8cddc0d9e84a3ae66d84b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-25 21:59:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d184f09f3f6d52203351c3a7ad9b173fcc301d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 23:33:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/aae3682cea894467b27e1ab956c9dd6d : SUCCESS in 55m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d5fe73cb9d8846f0a5ea57fee874cfab : SUCCESS in 1h 07m 58s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ca7a54a4c904cbea0be113b2b9c480d : SUCCESS in 1h 31m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a498c4d9a03940b7912742a584ecf8fb : SUCCESS in 53m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b487b77e5cd4700a7074a7b1da92c0f : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af1a821d85364d278e0fbd2c1e29564a : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a577aa52d32e4619b4bb57b5cec17a23 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da71d92f9be64bc28b1b723a4e4a458b : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0433242acacc4b13a8ec240643047d6c : SUCCESS in 6m 45s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/ab8ffe03757d454a85c67f45f3445853 : SUCCESS in 11m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/49b22a0bc206438aaf4e3eec2b7e4f42 : SUCCESS in 9m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/408d2035e2ff40ca8a9c61457ee74d01 : SUCCESS in 10m 26s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c8f9072548e24f63a39155bee82c591b : SUCCESS in 5m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f85f9cae160949ab9933c13397a68318 : SUCCESS in 1h 23m 36s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1e9a82d43577436b932a9929116b63b5 : SUCCESS in 1h 22m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ec2e31f02b664adab071cf4dcb1e495a : SUCCESS in 1h 26m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/10ab7e38c524404f9f4a31650ffc7b5e : SUCCESS in 1h 17m 31s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/213a6343378d4e24a8102a92dc4f4583 : FAILURE in 39m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b078495a194435ca68e2fe34e552ff2 : SUCCESS in 44m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a42d7fa063ff49609276b70160e42b66 : SUCCESS in 1h 26m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e3d10a54f6a4453988b414a34eff2214 : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":2},{"id":"fea61251d6b0a33575279b9aca1fc66848db84b4","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-05-26 02:31:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"351677c3ee663eb5650e0686135ee2e9aeb1a499","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-26 06:13:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1fbde21bd952974ae269f39c604a06ba49b78819","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-06-02 11:57:25.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8d0a2e3b92ae60a78f5053316906c368285a2ed","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-03 11:23:30.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69173bb3b50bd6503d83e6adc48bf9b9f27e7ea4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 11:24:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ba4926ce2926a928246bfa2b9ce50fc156cbb35b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 12:51:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/65d70a1cade74c4f84e65c51b1fbb746 : SUCCESS in 50m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/334ca180783c4b6bab8e04356e5104d9 : SUCCESS in 1h 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba25a5c0fd6f471fbe26a1c3dbf88215 : SUCCESS in 47m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1f8615353e749eab2c5f324e73bd5d1 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9633c03690d437fa43807f2ce46fb26 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/347ffcb3286844198967b515394e4001 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/769f2ce3104145d68b3dc073b5dfc20f : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f763d74b34844662a7c52b289c4f89a8 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/61cdd962d12945ebb8b03c0823faa89f : SUCCESS in 10m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/80cdf88d8837463581245af0dffd95bb : SUCCESS in 9m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cb455474699344b0b6a45d8b9d8f6cdf : SUCCESS in 1h 21m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/932ec77bb8474f78bd98ec176c2e53a5 : SUCCESS in 1h 18m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5f02ea574914062a1a583c75685aee7 : SUCCESS in 51m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/600c1699d7b84c4ea313dd22235d60bc : SUCCESS in 1h 21m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5068e7ddd69c475abc0bd76c022520a7 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":2},{"id":"cb0540b1293ba925d9e6475518db787fb917d080","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 12:51:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5f69caec0dc19abb02e748a20645513906a34a6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 12:52:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58606331b0d34952b73cdbe528a303f8 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d265c3fa85ca4d4fb76b30b6ad394174 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55c000ee1d7bb415491071abbc4001b4b715d380","revisions":{"c088c8b71f4b8e9e32681888e1333a1708d4339b":{"kind":"REWORK","_number":1,"created":"2022-05-25 15:55:35.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/08/843208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/843208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/843208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/843208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/843208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/843208/1"}}},"commit":{"parents":[{"commit":"efff2363e9bb22e4f32757dd278a70d0a9ed882e","subject":"Merge \"[CI] Add upper constraints to install command\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/efff2363e9bb22e4f32757dd278a70d0a9ed882e"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-23 20:34:52.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-25 15:55:35.000000000","tz":0},"subject":"Bump Image API version to 2.16","message":"Bump Image API version to 2.16\n\nThis API version increment is due to change I6a875a38bef5, which\ncorrected the response code for the admin-only API call:\n    PUT /v2/cache/{image_id}\nThe call now returns 202 upon success (it had been returning 200).\n\nThe intent of this version bump is to get the attention of operators\nto take note of this change; from now on, the call always returns 202\nupon success regardless of what API version you request.\n\nAdditionally, cleaned up some version tests that had gotten out of\ndate and were not thoroughly testing version negotiation.\n\nAlso, corrected the version negotiation code around the v2.13 API.\nThe versions response had been modified by change I6882fd2381e6 to\nonly include v2.13 when the \u0027enabled_backends\u0027 config option had a\nvalue, but the versions negotiation code was not updated to reflect\nthat.\n\nChange-Id: I0cf35ed7e21497826cd581e59aa58774f030b9f6\nRelated-bug: #1971521\n(cherry picked from commit 11989a18f4ede07026a435abb3d695a075ae7004)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c088c8b71f4b8e9e32681888e1333a1708d4339b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c088c8b71f4b8e9e32681888e1333a1708d4339b"}]},"branch":"refs/heads/stable/yoga"},"55c000ee1d7bb415491071abbc4001b4b715d380":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-25 19:38:45.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/08/843208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/843208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/843208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/843208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/843208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/843208/2"}}},"commit":{"parents":[{"commit":"9043efc5fff4daf83fdbcba1e200ab2e113e0464","subject":"[APIImpact] Correct API response code for PUT /v2/cache/{image_id}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9043efc5fff4daf83fdbcba1e200ab2e113e0464"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-23 20:34:52.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-25 19:38:45.000000000","tz":0},"subject":"Bump Image API version to 2.16","message":"Bump Image API version to 2.16\n\nThis API version increment is due to change I6a875a38bef5, which\ncorrected the response code for the admin-only API call:\n    PUT /v2/cache/{image_id}\nThe call now returns 202 upon success (it had been returning 200).\n\nThe intent of this version bump is to get the attention of operators\nto take note of this change; from now on, the call always returns 202\nupon success regardless of what API version you request.\n\nAdditionally, cleaned up some version tests that had gotten out of\ndate and were not thoroughly testing version negotiation.\n\nAlso, corrected the version negotiation code around the v2.13 API.\nThe versions response had been modified by change I6882fd2381e6 to\nonly include v2.13 when the \u0027enabled_backends\u0027 config option had a\nvalue, but the versions negotiation code was not updated to reflect\nthat.\n\nChange-Id: I0cf35ed7e21497826cd581e59aa58774f030b9f6\nRelated-bug: #1971521\n(cherry picked from commit 11989a18f4ede07026a435abb3d695a075ae7004)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55c000ee1d7bb415491071abbc4001b4b715d380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55c000ee1d7bb415491071abbc4001b4b715d380"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
