)]}'
{"id":"openstack%2Fglance~844596","triplet_id":"openstack%2Fglance~master~Ia0e5f52f3841b3306a8776762d18a56c6df1e2f5","project":"openstack/glance","branch":"master","topic":"bindep","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-06-16 16:56:59.000000000","reason":"Change was submitted"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2022-06-16 11:03:43.000000000","reason":"Erno Kuvaja replied on the change"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-06-16 16:56:59.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-06-16 16:56:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0e5f52f3841b3306a8776762d18a56c6df1e2f5","subject":"Bindep fixes for SUSE-like systems","status":"MERGED","created":"2022-06-03 14:06:59.000000000","updated":"2022-06-20 06:46:57.000000000","submitted":"2022-06-16 16:56:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844596-bindep","meta_rev_id":"32931de1c0aec77758c2ea9ae7b935832c725052","_number":844596,"virtual_id_number":844596,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-16 16:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2022-06-16 11:03:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2022-06-16 11:03:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-03 15:46:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-06 05:54:52.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-06-06 07:40:18.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-06-06 14:15:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-06-06 14:15:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-06-06 14:15:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-06-20 06:46:57.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"14c6e1fd671ff6ef7f2a9de5c9500ffc23fa764c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-03 14:06:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c94f764b5a9e111d9fe3dffe795cd1b2a30cc84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 15:46:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6ad38045b7cd43e9b96e4b92074006d0 : SUCCESS in 5m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/efab07915eba4c46bf8192297e8b3b9a : SUCCESS in 1h 02m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dc1f548901544c8d820ba4ef9b25122c : SUCCESS in 1h 22m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32fbc65cf7424d2c99a4dd7c9d4858dc : SUCCESS in 48m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40eb20fcff8645899c42d6b508a06953 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34b29f4d7bd640ed9a3fc9723f3dffe9 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f39ab3a100224f72815018f572c2c234 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b421f64ba8a47b69491fed761b6ca2b : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de6e917c4f641569d4c2efdeca8de30 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42f37129b8f4436d8eda470868b2383f : SUCCESS in 7m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9c4c72b133e141deb69dcc9d68f869b7 : SUCCESS in 11m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd97d5f752ce420ebd2716c335b68775 : SUCCESS in 10m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9e84de5aaf5d476f900f55e0327509d9 : SUCCESS in 9m 08s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/efd3c52429da4217b3192bc22ae28c51 : SUCCESS in 5m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e7d091bab3d94703b7b49d436995f9a0 : SUCCESS in 1h 14m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/409ae31251b142beb10816d3477d8f43 : SUCCESS in 1h 11m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/64fab3cbd26e4660a1300ca9816912a8 : SUCCESS in 1h 33m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/41484d6aa56f42ffbedcbfbcf4fffa30 : FAILURE in 1h 21m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01e0c6d25f5d4509bca567295704e32d : SUCCESS in 59m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5a639a4f30954515952692b7fd6db06d : SUCCESS in 1h 00m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c29a56b15d064831b55f02447764f17b : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a22df80bb4359d4d7d9fa245ef4e5ba213f1d75d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-06 05:54:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b2543a629e4fc321ba7372c8f339b01514d00e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 07:29:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d9284ca7a99e448099c4d03268d80a98 : SUCCESS in 4m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/22b718bc1ed040b89a7cbde7c7be288b : SUCCESS in 49m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8aa03a696cb0459fb294f5589d1aad5c : SUCCESS in 1h 19m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fdb4566f47e44bf9d3d342f0ad80ba7 : SUCCESS in 41m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9d5ca5165a742cea57ef30b38273e6b : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58cde4263e374d81948beb193c27a022 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d83854c2643845aea966329c26f6f336 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/561c7acf17284cd8bc7133b9758c6853 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b01d54fa36234ab5ba015dc46ede6b13 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93e717222ad343f99c86dc732511a890 : SUCCESS in 7m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/528dbabd1ac74a1abffffd6a1993eb85 : SUCCESS in 11m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/30ddd0519ce04574ba5ff856a1b74773 : SUCCESS in 7m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f115b7b0929e49f8a9a3b29166873593 : SUCCESS in 9m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1c7f7cfc2a5344b4ad1cb3ac2e212af0 : SUCCESS in 4m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5a687ea7ac7f4638a96d259d74a39bef : FAILURE in 1h 19m 48s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5f90e0bfa9654df1ad8779f24ea24250 : SUCCESS in 1h 06m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/54d830331f294cc0bedbd612087e2cf3 : SUCCESS in 1h 14m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bc29a2d053df4881bae811d9afc7acba : SUCCESS in 1h 33m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/645a09e88d7d477bb926cbb92d0d0160 : SUCCESS in 57m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7f502f86ae7642c382e51cffc0dc7a31 : SUCCESS in 57m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c73008b7df684745aacb28d99bfdc923 : SUCCESS in 28m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e0e1be22157c110ae4e78fa39d08b88e62f3b08e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-06 07:40:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3888ce8309fe94d82aa761bebcbf47fc35aa717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-15 21:09:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"641628b62df47c0149e17f39f41e4c9f214eb6db","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-15 21:09:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aaef7863bb8dcd574c5e2f133422e8382b8cdaf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 23:20:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e57e66525c4f454d9608275b0ff86578 : SUCCESS in 5m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/188091fd94f64680a79b8e4ff83ec44a : SUCCESS in 53m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d90c8c5cadb949c3b268de77bb129924 : SUCCESS in 1h 06m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69cb65457a944c31bc9ef1cacb6c3d92 : SUCCESS in 36m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88f07afa2eeb45dfb94f4130050c6c6b : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6473cf7247a4a159f716be7469e439e : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa31c996602f4e849f7d1af9d6419b84 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b549361cf68f46d48f97a3b6b04c4d38 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9486238175e24cb4896c574fbd21b2a1 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8d0da673c944d859cb98adece0a8abc : SUCCESS in 5m 46s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/76cb064b04a141f09a92138c48ab604e : SUCCESS in 10m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cc0353c3d9924c878077aafd3121c399 : SUCCESS in 8m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d0a2667bdebd4440968b760ea0b4dd91 : SUCCESS in 9m 25s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9c695e418fb34b3382bf131a37c8bce7 : SUCCESS in 4m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b6ec6b9c88364f94b8a2182474d51d87 : SUCCESS in 1h 05m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/73833d013e8d42e4a4538afc8fe33151 : SUCCESS in 1h 01m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f1ab6f352f85452db0572d6ea9d364a0 : SUCCESS in 1h 30m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3c73d0d7735e4248a78dc09bdb1aa635 : SUCCESS in 2h 05m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e15e4a641fb42f0a1160cd1ce5f0de4 : SUCCESS in 39m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4fd2f661031d40ef902ee6d5f5237031 : SUCCESS in 1h 15m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/94a004d1306442e99de1b1f86a12e1f6 : SUCCESS in 28m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ad5c955ce057b30ebd299fb8c59f4103cedf6e3d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-06-16 11:03:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe4e1a0b66e8d4c1adc53125fdf7794ca7e32ab3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 11:04:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c262be92f4b9e349cc661b6747e7f12164f4b01d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 13:31:17.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4d793a8fea4e46a39a3615e41b231132 : SUCCESS in 4m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/5125d9f1315043ab991ac47f7e0409f1 : SUCCESS in 56m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7830eb5ba9d74a4eac7cc660efbd7003 : SUCCESS in 1h 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52008b51739f471aafe5c7f54d363845 : SUCCESS in 48m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d00297c085a4460b115b1ead7426701 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7523c731b144854a9d75dadcb5546c4 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2e251402f634caeabae707fa0ede561 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d54e8930735a4ae5ba268849372b541d : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/859ed0568de44fa48e9baedbd692fc2f : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8364b47bb08b44f796b8e882d3af91dc : SUCCESS in 8m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3a6347368fcc4dd1ac147c22f4d864fd : SUCCESS in 8m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/288f00b71d0e479b8efd1756bb8d1102 : SUCCESS in 1h 11m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/143e6f8acb394c289e214d1be36ac10b : FAILURE in 1h 06m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2300021139e44b2b82ab4b11367e744 : SUCCESS in 52m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c66dd56fd355469e9d1203a5fe0208bb : SUCCESS in 1h 10m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b4f06a8160bb4f38a053cb2c868613f9 : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f91aeb7ee678676bd5615d218dbae0853fabcff1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-16 13:47:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c5d5b094fca2656b8bf8a81cca3e6a88c8c16fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 15:30:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4131d9f8c757432294f6bb15376aec94 : SUCCESS in 5m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/b90b748cc03f4d919e83e51ce93c57d9 : SUCCESS in 1h 16m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/249dacba2e334405ab0d812d1c739b56 : SUCCESS in 1h 14m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5817cf5d31fc4867a3a98791dc443b1e : SUCCESS in 55m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62681c7c65e74c919d1f1ec37f7307ad : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e3248ad43f848c282ea96d17e12cb13 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/991c914ba9f24f329a5215530bd4f75a : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d969adf0ae424ef6877e8da9b27d443a : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab4343ba67d343438285f3d45720b8db : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a518394a0e574bed9af3edf7d58b4f78 : SUCCESS in 6m 25s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2dd1c05372fd455789bc34e57adfb618 : SUCCESS in 12m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9faf637d9c624e64ace55bbfd4acc620 : SUCCESS in 10m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7309f715afa7421dbd8e53fb38b3b68f : SUCCESS in 10m 51s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0cb0d337d89b4f71b2a41a31e90404c0 : SUCCESS in 5m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/105e504ef4cb4a7d86f41f79b8b0c60e : SUCCESS in 1h 32m 20s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bb6ae99d54eb4427a8090924904a2b97 : SUCCESS in 1h 19m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/024e62126ed44780a0fba02bb77a4cdb : SUCCESS in 1h 33m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/63bd9f0c08bc49dbbc80c780578229f5 : SUCCESS in 1h 34m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acdad15d24a1411089ae7191165ce014 : SUCCESS in 57m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3aa6a7159e4c44a0b60368b678ef8177 : SUCCESS in 1h 05m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/acd452e933f644a6a91282ba6ac167c8 : SUCCESS in 33m 08s","accounts_in_message":[],"_revision_number":2},{"id":"af08a96c6ff28438f635227d45031c2ab940c4e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 15:31:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6e2737391f3d87e5dfa33f2378ec293c61b0879c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 16:56:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1992db94164049d989b47f6842a6966f : SUCCESS in 5m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/6b02fe43e87a493eaa45d455917faab8 : SUCCESS in 53m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3564e4db4e3d44189012711e472bb390 : SUCCESS in 1h 03m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cd28065763b4e798da5940439077f8f : SUCCESS in 42m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be0e8951db764950b73e5db78423bad8 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ac84feeb73c466f9086839188fd6d2f : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67e1f78be8794bd2a14f2a38bbcd67bb : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43f7a58a44cd4c91850d6a2dfe589ac7 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d963e3f747f74617b34fcede0256729d : SUCCESS in 5m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0cb58d8f0ef54fb594c45637384c2730 : SUCCESS in 9m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/753681a4d1fa44d0a1845d9f18efba99 : SUCCESS in 9m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/312630e6b9ed49179e687ab1d85679c4 : SUCCESS in 1h 20m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/edc3088ec4ad44b79a07721c46663f4d : SUCCESS in 1h 05m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d56598353b3142dfb8fb1a517ece9788 : SUCCESS in 50m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0b5898e7cd4e4bc39e25523d3c0e965c : SUCCESS in 1h 14m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c1918332c73d490ba58d2e0585bdfb2c : SUCCESS in 35m 49s","accounts_in_message":[],"_revision_number":2},{"id":"d9c231050c36a2f35fe46e634d5bedf969756914","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 16:56:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8a6fd23ea458970b39a63ff287a91f7a209e2bbb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 16:58:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/7f24af68bff0410eaa8bc313a50946f8 : SUCCESS in 55s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d632f8c480094a4bbabf988d729d8c7c : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e61ad041b1c14b0cba07b69ece6ea2a5 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2},{"id":"32931de1c0aec77758c2ea9ae7b935832c725052","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-20 06:46:57.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/844596/844596/2/check-not-vote/ext-glance-zuul/dd97a22 : FAILURE in 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53f322f1d8c0a5c1e8c494f88f1f05c47b29729a","revisions":{"a8c077f1d3c9afd47b95aa1db7b13a16e164e6b0":{"kind":"REWORK","_number":1,"created":"2022-06-03 14:06:59.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/96/844596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/844596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/844596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/844596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/844596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/844596/1"}}},"commit":{"parents":[{"commit":"6bd1d14a528e8454d83434e4d7e9bd8fc28a4390","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bd1d14a528e8454d83434e4d7e9bd8fc28a4390"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-06-03 13:49:20.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-06-03 13:49:45.000000000","tz":120},"subject":"Multiple bindep fixes","message":"Multiple bindep fixes\n\n- libpython3-dev is not required for dpkg-based systems\n- on SUSE-like systems, qemu-img is provided by qemu-tools\n- on SUSE-like systems, pg_config is provided by postgresql-server-devel\n\nChange-Id: Ia0e5f52f3841b3306a8776762d18a56c6df1e2f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a8c077f1d3c9afd47b95aa1db7b13a16e164e6b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a8c077f1d3c9afd47b95aa1db7b13a16e164e6b0"}]},"branch":"refs/heads/master"},"53f322f1d8c0a5c1e8c494f88f1f05c47b29729a":{"kind":"REWORK","_number":2,"created":"2022-06-15 21:09:18.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/96/844596/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/844596/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/844596/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/844596/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/844596/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/844596/2"}}},"commit":{"parents":[{"commit":"ebef97d776eb1982b948658eaf20566853dc70a9","subject":"Merge \"Modify wrong file path in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebef97d776eb1982b948658eaf20566853dc70a9"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-06-03 13:49:20.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-06-15 21:07:06.000000000","tz":120},"subject":"Bindep fixes for SUSE-like systems","message":"Bindep fixes for SUSE-like systems\n\n- qemu-img is provided by qemu-tools\n- pg_config is provided by postgresql-server-devel\n\nCloses-Bug: #1978869\nChange-Id: Ia0e5f52f3841b3306a8776762d18a56c6df1e2f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53f322f1d8c0a5c1e8c494f88f1f05c47b29729a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53f322f1d8c0a5c1e8c494f88f1f05c47b29729a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
