)]}'
{"id":"openstack%2Fglance~845511","triplet_id":"openstack%2Fglance~master~Ic495c0e58b9d471da64eb0826889d654716927e0","project":"openstack/glance","branch":"master","topic":"bug/1962480","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-06-13 10:48:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic495c0e58b9d471da64eb0826889d654716927e0","subject":"Update api-ref: versions response for master branch","status":"ABANDONED","created":"2022-06-12 23:45:54.000000000","updated":"2022-06-13 10:48:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"399db47a07c01c63fa9da76dbc4b019dcb4bfbd1","_number":845511,"virtual_id_number":845511,"owner":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-13 00:11:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-12 23:51:17.000000000","updated_by":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-06-13 00:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bd4916a09630697e9a195c631945793542bbef5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-06-12 23:45:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a30705ff3e0ec4f1530eaf77416a0ce25c498f70","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-06-12 23:51:17.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"102a2cd5777ceca1bf9f56003f291d4ec28d939a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 00:11:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bc8c65823e2f4551bcae2707947b7013 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a554f68b10ff4fab8bbf328728210f68 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f86ab36c3984ff6a2eef3ae4d93d347 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"399db47a07c01c63fa9da76dbc4b019dcb4bfbd1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-06-13 10:48:59.000000000","message":"Abandoned\n\nThere are other patches tracking","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d48737852a4d7a42812ad8722e7669c6aeb0153","revisions":{"3d48737852a4d7a42812ad8722e7669c6aeb0153":{"kind":"REWORK","_number":1,"created":"2022-06-12 23:45:54.000000000","uploader":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"ref":"refs/changes/11/845511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/845511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/845511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/845511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/845511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/845511/1"}}},"commit":{"parents":[{"commit":"ebef97d776eb1982b948658eaf20566853dc70a9","subject":"Merge \"Modify wrong file path in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebef97d776eb1982b948658eaf20566853dc70a9"}]}],"author":{"name":"chenwei","email":"chen.wei@99cloud.net","date":"2022-06-12 14:50:45.000000000","tz":480},"committer":{"name":"chenwei","email":"chen.wei@99cloud.net","date":"2022-06-12 23:40:59.000000000","tz":480},"subject":"Update api-ref: versions response for master branch","message":"Update api-ref: versions response for master branch\n\nCloses bug: #1962480\n\nChange-Id: Ic495c0e58b9d471da64eb0826889d654716927e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d48737852a4d7a42812ad8722e7669c6aeb0153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d48737852a4d7a42812ad8722e7669c6aeb0153"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
