)]}'
{"id":"openstack%2Fglance~847089","triplet_id":"openstack%2Fglance~master~I795c751edd1403d23aa1de5cda194aada80d05b2","project":"openstack/glance","branch":"master","topic":"fips-compatibility","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-06-23 13:43:08.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-07-01 14:19:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I795c751edd1403d23aa1de5cda194aada80d05b2","subject":"Fix fips job","status":"MERGED","created":"2022-06-21 19:38:18.000000000","updated":"2022-07-01 14:22:38.000000000","submitted":"2022-07-01 14:19:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847089-fips-compatibility","meta_rev_id":"3742996e90342e67f55b5f697d6227dcfc6a835d","_number":847089,"virtual_id_number":847089,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-01 14:19:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-07-01 06:24:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-07-01 13:33:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-07-01 13:33:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-21 20:20:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-22 14:31:48.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2022-07-01 06:24:57.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-07-01 13:33:51.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"404d4984804c2e8e261801e970ef859223abb40e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-06-21 19:38:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6284559fd12f58ef361783dabb5015aa93cb28a","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-06-21 19:38:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1334ed9f2b6b5a555e1d171b4e0392362cac05fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 20:20:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57b5b9751479455b99794373dc67d959 : SUCCESS in 39m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8bd602ca12c46d6bf368e704edbd437 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5b827397fe44215890947f3e187b575 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01bfca0fab934b7dba774868fbf86174 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c5148af7b834e0f8e1b318987bdbc02 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e3ee23fe7874b25b8048009d7937de6 : SUCCESS in 5m 46s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a27e712deeca4228a6813dd168e69845 : RETRY_LIMIT in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/86f9e80ca371494aacf75a8d703ae0e8 : SUCCESS in 8m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1d3def7886684b1585623197275beb89 : FAILURE in 8m 31s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/356441f2dc65455aaaa5b9e8d451df4f : SUCCESS in 27m 36s","accounts_in_message":[],"_revision_number":1},{"id":"965c8822c6fd71c6576433ce7a6b0164ff0fe9c3","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 21:16:09.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- glance-tox-py39-glance_store-tips https://zuul.opendev.org/t/openstack/build/f1a9dc3f953c42c8ace563abb293e89b : SUCCESS in 6m 00s\n- glance-tox-py38-glance_store-tips https://zuul.opendev.org/t/openstack/build/12114f65ceca441a99000ee25c6d30cd : SUCCESS in 5m 38s\n- glance-tox-functional-py39-glance_store-tips https://zuul.opendev.org/t/openstack/build/2925f2a7d6d44a54bdc3fe936158ca5e : SUCCESS in 10m 25s\n- glance-tox-functional-py38-glance_store-tips https://zuul.opendev.org/t/openstack/build/1aa9ab70ed68471b884c6dce4a4c6baf : SUCCESS in 9m 33s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7e8f97d58d404eaa9063ee3da0e501a4 : SUCCESS in 37m 44s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/8985d405fa7f46a688e14473d7764615 : SUCCESS in 1h 05m 41s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/1d6fff32c63348f09489fdd78c0a079f : FAILURE in 1h 36m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e28b54d5de361d1eed11bf007ac4854ba9143e75","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-22 14:31:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e645f34bc38b000760e2ed8a84657286f5735e2","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-06-22 16:17:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05db379f67651a744badc1c1b2f118f372fc2e6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-06-22 20:49:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9b2ed05ab77d1d00a080a81cb03de29052e1a836","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-06-22 20:50:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"227a1b7995c638d0d4891bda8a4ecf76784515c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 21:56:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3942139a0c94ec8b2fe0dad0accdcdf : SUCCESS in 44m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa597f86f8744f6d94063cb423e0c4b5 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b096c7392f34f60abffc0e27f500eea : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c56282fdfbe742bda58cec6f17edc27e : SUCCESS in 6m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30f8d13cd24d4400afd47390e1f33108 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32f95f2d0a9437696bf8e9bbc3a56a2 : SUCCESS in 7m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fac63f885b4b45b992843b7c522dd3d6 : SUCCESS in 11m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f96d21ca2f947ffb4b4a34a7f2e7819 : SUCCESS in 9m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5b6b0fc061a548c992e29480de85b9a3 : SUCCESS in 9m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c743b30d6c864118adb8e4b576f64014 : SUCCESS in 26m 19s","accounts_in_message":[],"_revision_number":2},{"id":"3f2fa226f41cc873fc4bbcd1d5b32222ec07f6ad","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 23:03:11.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\n\n- glance-tox-py39-glance_store-tips https://zuul.opendev.org/t/openstack/build/c7a8788b2ce14ee1853693a3e0e4f735 : SUCCESS in 6m 03s\n- glance-tox-py38-glance_store-tips https://zuul.opendev.org/t/openstack/build/09f6f6e23bcd425fb2712b95dcdcab7e : SUCCESS in 5m 50s\n- glance-tox-functional-py39-glance_store-tips https://zuul.opendev.org/t/openstack/build/a625b6bb80834263be328a4395f9e940 : SUCCESS in 9m 27s\n- glance-tox-functional-py38-glance_store-tips https://zuul.opendev.org/t/openstack/build/26566cf97a85400680d685d4e78339e6 : SUCCESS in 8m 43s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9b13644bbf484fda875e299f7f3f4203 : SUCCESS in 42m 38s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/78dbe813b9a94345bc192a79145da18a : SUCCESS in 1h 11m 56s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/47bbd9fe80e0435b8a495eca9e735d15 : FAILURE in 2h 06m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ca67819b2f7f0baefbed33120968fe89c36d6c6c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-23 13:43:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"516b77aac5b4bff570c8735038ae82b584755f03","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-30 14:50:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cef04fbbcb6592398bfc356d6642e75e274bc62e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 16:58:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- glance-tox-py39-glance_store-tips https://zuul.opendev.org/t/openstack/build/ba29218d3aeb4500b1235c59c7eebca8 : SUCCESS in 6m 35s\n- glance-tox-py38-glance_store-tips https://zuul.opendev.org/t/openstack/build/b8db488075a44658847f9d3e254893f1 : SUCCESS in 6m 00s\n- glance-tox-functional-py39-glance_store-tips https://zuul.opendev.org/t/openstack/build/ff73a95c35784076a99451904e1154c9 : SUCCESS in 9m 43s\n- glance-tox-functional-py38-glance_store-tips https://zuul.opendev.org/t/openstack/build/f346d0ff648e4a8bacf4b9aac370476b : SUCCESS in 8m 41s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/541b1b5c80ba4a6588a717a0da18b8d8 : SUCCESS in 1h 10m 49s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/6a5b892bec044f67b439e6779d7262c3 : SUCCESS in 1h 13m 23s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/126d8942a719447983236561adca096b : SUCCESS in 2h 02m 31s","accounts_in_message":[],"_revision_number":2},{"id":"eaac013e91648e68366212e85ea471aeb225d401","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-01 06:24:57.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3684e77d273745cf2916ec668b781dae339329e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-07-01 13:33:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2489d8b20131e3620e690f57ce0c628f4158d778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 13:34:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f274944000c23a5f605fc327779ee415258fc48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 14:19:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc826c05056c4e7ab91d5c63f071d2d6 : SUCCESS in 44m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8119331ba00547ba8b8430a00a936fc9 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01c93f4ad49b41ef8afb154f337c9c11 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a10ed57743fa4ff6b7a62df4f39a52a0 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79a4ac60ed34674b6d22cd806690482 : SUCCESS in 8m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c7557e062b0a4dbe952e5d78f9a394a1 : SUCCESS in 10m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6114fe6f1836460786319c845a686d3e : SUCCESS in 10m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/58507b5d64d343ecacc53d772f91b8e4 : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":2},{"id":"8d8309220bb623663a6177e5edbf1f574c99f0be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 14:19:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3742996e90342e67f55b5f697d6227dcfc6a835d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 14:22:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8eeb968c042490f967be0ea490fc565 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"05543914a774bb4a170b789f64e6945bcfbc4259","revisions":{"97e446b32b2cd7672014753f6245848b4c48c17c":{"kind":"REWORK","_number":1,"created":"2022-06-21 19:38:18.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/89/847089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/847089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/847089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/847089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/847089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/847089/1"}}},"commit":{"parents":[{"commit":"53f322f1d8c0a5c1e8c494f88f1f05c47b29729a","subject":"Bindep fixes for SUSE-like systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53f322f1d8c0a5c1e8c494f88f1f05c47b29729a"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-06-21 19:36:29.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-06-21 19:36:29.000000000","tz":-240},"subject":"Fix fips job","message":"Fix fips job\n\nThe ecdsa usage for ssh is supposed to be the default right now.\nAlso, a new role paramter nslookup_target is required for the fips role\n\nChange-Id: I795c751edd1403d23aa1de5cda194aada80d05b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97e446b32b2cd7672014753f6245848b4c48c17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97e446b32b2cd7672014753f6245848b4c48c17c"}]},"branch":"refs/heads/master"},"05543914a774bb4a170b789f64e6945bcfbc4259":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-22 20:49:38.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/89/847089/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/847089/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/847089/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/847089/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/847089/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/847089/2"}}},"commit":{"parents":[{"commit":"53f322f1d8c0a5c1e8c494f88f1f05c47b29729a","subject":"Bindep fixes for SUSE-like systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53f322f1d8c0a5c1e8c494f88f1f05c47b29729a"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-06-21 19:36:29.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-06-22 20:49:23.000000000","tz":0},"subject":"Fix fips job","message":"Fix fips job\n\nThe ecdsa usage for ssh is supposed to be the default right now.\nAlso, a new role paramter nslookup_target is required for the fips role\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/847193\nChange-Id: I795c751edd1403d23aa1de5cda194aada80d05b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05543914a774bb4a170b789f64e6945bcfbc4259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05543914a774bb4a170b789f64e6945bcfbc4259"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
