)]}'
{"id":"openstack%2Fglance~847139","triplet_id":"openstack%2Fglance~master~I01e8550aef85d333a70392be83a964267e632c2d","project":"openstack/glance","branch":"master","topic":"bp/cache-api","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-07-28 15:53:11.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-07-27 20:27:20.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-07-28 14:26:48.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-07-27 20:11:05.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I01e8550aef85d333a70392be83a964267e632c2d","subject":"Enable Image cache tempest testing","status":"MERGED","created":"2022-06-22 09:18:55.000000000","updated":"2022-07-28 15:55:38.000000000","submitted":"2022-07-28 15:53:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847139-bp/cache-api","meta_rev_id":"30c5c7bf5ee98e8f52f2a77af38b1e8bad113e4f","_number":847139,"virtual_id_number":847139,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-28 15:53:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2022-07-27 20:27:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2022-07-28 14:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-27 20:11:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2022-07-28 14:26:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-22 11:18:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 07:49:07.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 06:04:47.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2022-07-27 20:07:25.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-07-27 20:07:25.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-07-27 20:11:05.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"234e0d014f7b46bc6b610200ab5f8d336bb8ce0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-22 09:18:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"686861624c05fa03f45f9d74c1810f7e7bc9471c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-22 09:20:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"737ae5e6c985230647134d0736cfe28bfe4102db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 11:18:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e62f89345984be683a3fab57507062a : SUCCESS in 45m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fb6967208e34b538bfd29d0f9eb7762 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fd9c5d1df47438da52fe2bcac6529e8 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d74b25cb5cc4cdb815b105b4151a80b : SUCCESS in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69da15a496804c83afbab5ff075f2b07 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef5ef7bb8024e21beaaaf9efccd1902 : SUCCESS in 6m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fcb2d92e32824875893d7de011c79e00 : RETRY_LIMIT in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e7a1e04305fb4c6cae1334f6a62da766 : SUCCESS in 8m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5484fc4ea0fa46eb8e0cd2283cedd297 : SUCCESS in 8m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c39e87fb856c4a138546147610e7b787 : SUCCESS in 1h 15m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/24d409a1797544f7972a4e5b22883fa5 : SUCCESS in 1h 21m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cce0d278fc61473bbf0f9ff7a451dcbc : SUCCESS in 1h 50m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a7984f53d1c34ba28e0287be7bc33031 : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":2},{"id":"317efa931bbbfcae4f8151004ef52d73e6eef58d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-22 16:15:23.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2110e077d8c32a1b465cec6eb77af11147ad151c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 17:55:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64b5821068f64010b1ce37fc1298caa5 : SUCCESS in 38m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec9eedecd736437491e60604f7e02d53 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7105fa2263954840800605aa317b55c0 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54478be24c8b46b7a73a5c17c891c415 : SUCCESS in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3034647c02414546a86dadbb09eda924 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3830ae76d81f469e9b2b1abef6e043d2 : SUCCESS in 9m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/80c7193adaee4a0aa9cc109f1f72e13c : SUCCESS in 10m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/734875b6e1a94899aa97a13c762b57b8 : SUCCESS in 9m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/01c0f610acb847bf9a14814509900833 : SUCCESS in 9m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/07ccbbab781c4c2394488f1ac62efad9 : SUCCESS in 55m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5eabb19605b24a31b8f0aa855edcdd7c : SUCCESS in 58m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3112c05db6ac4ca481f50ddc651d843a : SUCCESS in 1h 34m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/670f040f3cc84a4c9d061e267239dbbc : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":3},{"id":"dfb00d4df86ee704865ec1c85552d4b3ad16c9c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-30 07:14:09.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9d24cb2e13c7d37326967292bf472f7439f3fdd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 07:45:01.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"cdc4dc9a2ac1f0d8030b2881a783bd4480735366","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-30 07:57:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cca90bc5c619b34fd552bcdace341b796fdb90b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 09:50:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41217c1600734c80a19ac1d0f3d23670 : SUCCESS in 51m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a0c44cf200441968c8660ca8b46dfaf : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9d3c67d851e4a4fb231171b31a572e7 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7ff11ca8a5b44ebb870fd710caf86e6 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71ead7e5d94b43a9b184c3813e6bdf4b : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dbf0f32c12e4eb09938fe345e4f6225 : SUCCESS in 6m 43s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8c548221dc164b38a3f9d57e621ba19e : SUCCESS in 13m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b1ad1f1d4de148e49a5bf3ae33f172f8 : SUCCESS in 10m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5e71c10f58ff4f3b8495e5a0c864af1c : SUCCESS in 10m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a563007fe98f41cb99abee4f0e4b118c : SUCCESS in 1h 30m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8a93fefda08046eaa48ad6d77d0fecdc : SUCCESS in 1h 47m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8071a8c6ad4e4d63873fe61f5fdd72f1 : SUCCESS in 1h 25m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/81000ce4fe544ba3aaa7b84edc37a88a : SUCCESS in 40m 12s","accounts_in_message":[],"_revision_number":4},{"id":"355530a6bbc3cdff7e71c4e4f788b96eb77e4da5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-12 07:48:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5d4bed81b37644d42cb4a01b581c12be8b20d1fd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-12 07:49:07.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/847139/847139/4/check-not-vote/ext-glance-zuul/4e0c16f : FAILURE in 26s","accounts_in_message":[],"_revision_number":4},{"id":"7cfa7335756265d64a11879b94dc1217d9ab6c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:32:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/295eabb376ea45e680d7916ad1dcc510\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b041c25a804459faf3ae624c98e89b0 : SUCCESS in 54m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa0dab7b21642b1bd67b381765a11ef : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e61fc821cb44947b5753b9e6aa38649 : SUCCESS in 8m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/812569bdd4324970bae843a2c2620824 : SUCCESS in 9m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eadfd934b57c4c53bb91ad154f9084a9 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e5ed7eb44a54784a61f06d4e777e830 : SUCCESS in 6m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/889553190e1b4654b8937ef112f9f4cc : SUCCESS in 12m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/01f7784beadd481f909a25eab7c2ae8a : SUCCESS in 11m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3d28eb62d2994285bcb96aea6651c3be : SUCCESS in 9m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7ebcea9f6cdf4bc9a48d8213e110bbea : SUCCESS in 1h 29m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/084592f740e9463598ab5320054fd55f : SUCCESS in 1h 29m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/44771bef9c4c4ef1903e47603d917686 : SUCCESS in 1h 37m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/27bfe87295ec44ed9b1e91042fc09b77 : SUCCESS in 43m 21s","accounts_in_message":[],"_revision_number":4},{"id":"299c9c6954f5d642ca827745d43c11b020224dac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-20 17:04:02.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"6e41dfb74c5c5afdac7ccde5d20cc7511ce46652","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-20 17:49:27.000000000","message":"Topic dnm-tempest-coverage removed","accounts_in_message":[],"_revision_number":5},{"id":"317288a68c2f22e870cfaf115e5a639f1610c06a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-20 17:49:31.000000000","message":"Topic set to bp/cache-api","accounts_in_message":[],"_revision_number":5},{"id":"8a31f229067e6347d927b820b8a2fba1bfddae98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 18:52:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9e7e456e2304b7ebafe1a9de7f0af25\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58a61d2100b147329d7ef4c0a73d8219 : SUCCESS in 53m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a64845745f24697b48419cd2ddbd8ba : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/690c2c2884054659b167a630bc88fa0e : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0738cb82bf03429f9e0a7eba37beb893 : SUCCESS in 7m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab8c8458a3484119aa6474362aa2e667 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7b2b8bc4a242a4873295e5f9ec1542 : SUCCESS in 7m 52s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/90db928953c042f7b5f7aa12d092dbb0 : SUCCESS in 13m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/50d7ea82522249daa654cbebb72a128b : SUCCESS in 11m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b218d612437f471c89ccfa1ee341830d : SUCCESS in 11m 03s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/81045d787f974c8fa69c0ca625814638 : SUCCESS in 1h 44m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b27e221e765a4b0e98685b05222cf186 : SUCCESS in 1h 29m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ada05037836b4f458c4b589fcab2853b : SUCCESS in 1h 46m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a03fc5dc70f84a73ad77d6ab24ff8e87 : SUCCESS in 46m 44s","accounts_in_message":[],"_revision_number":5},{"id":"bf7f2af7e23d074f2d605ffe0381c4329b22e3a9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-07-27 06:04:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c37b82db1c17a77966d1a6d192528e4a8e8cd175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 07:35:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/855316ebe98d48478018d35afa4f2510\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc7b85269cae4da1965133589bedad9b : SUCCESS in 56m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d0fe7114888417f969afa82804e0edb : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ca3cf04ff864e53b4befae81cbd2f24 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00635bcf5f244e7cb89c399bbe82a881 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad6cc41340ba49ffa30953c89991db1e : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd2d3bfb48c488e8a561835c08517c7 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4b64c43f6b8842e3985654f4df84adf0 : SUCCESS in 12m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6fe50994caa4452c990342d51985d8b4 : SUCCESS in 11m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bca1df54e8804dd9a5283967e28d03a2 : SUCCESS in 9m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d9cad3ca669e41bcb825ab364357481a : SUCCESS in 1h 14m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0a929b4adfab4cc0a587abc30afdc1fd : SUCCESS in 1h 29m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8b57e1fe3ff042099a5e6334dbbba265 : SUCCESS in 1h 27m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6e1c2fd0562247c59234843e7224b730 : SUCCESS in 26m 40s","accounts_in_message":[],"_revision_number":5},{"id":"839d934e3fdc27b5606dac62fdfce759e7d00605","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-27 14:11:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b28a6ca3cc3e005134a6844dbc5599c44018ee59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-27 20:06:55.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"0039ab1da1f70d9ae9d966f8a897f27c7b018a48","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-07-27 20:11:05.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"4d0d2f14f12999bdfa8ca3992a8805f0ddc049c1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-07-27 20:27:20.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"8110caa566f755d31bdc868a7ebfb3d621c0043f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 21:39:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e38a59c5497c4207a1fe650d266c0150\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ec10d2eb2d14391ae009c992af96e79 : SUCCESS in 52m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8014e1a58247418085a1a5c5774c46 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af3ed8a52fe645ee9bcb92cd61b833a9 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06434e536e204d42ab43367c21be701e : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/624e2922163d409baaa9464eb3a9effd : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a1b19c672814983971d74764a2074f3 : SUCCESS in 5m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/138cea54a865471e9df4e3447ea7ba6a : SUCCESS in 10m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bfc2098b17034527ae0d110d6f1d15e9 : SUCCESS in 8m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7312aa763ed545a0b38a7919b688c61c : SUCCESS in 8m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4712be9a94d3475ab906abcd66cea9b2 : SUCCESS in 1h 09m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/11392c4d5f70464bb45a26bf455e05b1 : SUCCESS in 1h 16m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e137408e05774bae99fe33ed3dbcb839 : SUCCESS in 1h 31m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1d9801b1c6a949e09211f9e380bce25b : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":6},{"id":"fcf666059ad2a0634ccd66c024783b558e8d1b8e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-28 14:26:48.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ee140bd33a2c49d9aa953905de3bd0ebee078e8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 14:27:30.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4b701e1ce80f698949f572af3985bbb2c56c53a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 15:53:10.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70b02812550846fe992a69accbe5c180\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95bfef0661a24271973e820d9070b214 : SUCCESS in 52m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b928abb624345cf9777ffcb832bd59c : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67cbef716fca45a2843df58fba44b274 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/203e010f916844c4875937faf4de4437 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5c9e1cce26f4e49afa707c1e37b9d73 : SUCCESS in 7m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/22b459c8a57440dca9f9927c2cea8f56 : SUCCESS in 9m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b39e847e0b2e46548f5922ec3c60f7a1 : SUCCESS in 11m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/255aee5500e04cd086dfed0c0dbec699 : SUCCESS in 1h 24m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a44e360572b74a69bbb9eb7c3c3be79b : SUCCESS in 1h 17m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3dd4eaceae5a4023b112677181743192 : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":6},{"id":"d1f10625f45e876723e376d47a4788122a30eed5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 15:53:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"30c5c7bf5ee98e8f52f2a77af38b1e8bad113e4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 15:55:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6dd215ddf3f4979b7fc6a5ffe558e31\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e90194d113e412dad3b66a6e7232d4d : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1ef06ef08ba177bd9ea93c25f96fa13aa6a2e9de","revisions":{"eae7d36414b5a7d0fc8e944c99833210cd747b89":{"kind":"REWORK","_number":1,"created":"2022-06-22 09:18:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/847139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/847139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/847139/1"}}},"commit":{"parents":[{"commit":"bf226965af00bb989aec55ae72ef210c5b153cde","subject":"Immediate caching of an image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf226965af00bb989aec55ae72ef210c5b153cde"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-22 09:13:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-22 09:13:00.000000000","tz":0},"subject":"DNM - Image cache tempest testing","message":"DNM - Image cache tempest testing\n\nChange-Id: I01e8550aef85d333a70392be83a964267e632c2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eae7d36414b5a7d0fc8e944c99833210cd747b89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eae7d36414b5a7d0fc8e944c99833210cd747b89"}]},"branch":"refs/heads/master"},"93bf1fb5b942d0a847d036a5f8237ba7ce08567c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-22 09:20:17.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/847139/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/847139/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/847139/2"}}},"commit":{"parents":[{"commit":"bf226965af00bb989aec55ae72ef210c5b153cde","subject":"Immediate caching of an image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf226965af00bb989aec55ae72ef210c5b153cde"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-22 09:13:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-22 09:16:28.000000000","tz":0},"subject":"DNM - Image cache tempest testing","message":"DNM - Image cache tempest testing\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/825112\nChange-Id: I01e8550aef85d333a70392be83a964267e632c2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/93bf1fb5b942d0a847d036a5f8237ba7ce08567c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/93bf1fb5b942d0a847d036a5f8237ba7ce08567c"}]},"branch":"refs/heads/master"},"458d49cfce1032f5b9ea3d95f6a8c0e7b188ddb0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-06-22 16:15:23.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/847139/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/847139/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/847139/3"}}},"commit":{"parents":[{"commit":"bf226965af00bb989aec55ae72ef210c5b153cde","subject":"Immediate caching of an image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf226965af00bb989aec55ae72ef210c5b153cde"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-22 09:13:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-22 16:15:23.000000000","tz":0},"subject":"DNM - Image cache tempest testing","message":"DNM - Image cache tempest testing\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/847193\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/825112\nChange-Id: I01e8550aef85d333a70392be83a964267e632c2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/458d49cfce1032f5b9ea3d95f6a8c0e7b188ddb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/458d49cfce1032f5b9ea3d95f6a8c0e7b188ddb0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"39215c4925ad4bb1f82809e5f66da006a39de2b9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-06-30 07:14:09.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/847139/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/847139/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/847139/4"}}},"commit":{"parents":[{"commit":"a71f7ba44e57a7945addb82afd2a19962d5c04cc","subject":"[APIImpact] Correct API response code for DELETE cache APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a71f7ba44e57a7945addb82afd2a19962d5c04cc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-22 09:13:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-30 07:14:09.000000000","tz":0},"subject":"DNM - Image cache tempest testing","message":"DNM - Image cache tempest testing\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/847193\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/825112\nChange-Id: I01e8550aef85d333a70392be83a964267e632c2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39215c4925ad4bb1f82809e5f66da006a39de2b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39215c4925ad4bb1f82809e5f66da006a39de2b9"}]},"branch":"refs/heads/master","description":"Rebase"},"a3b8a54ccd6674bb1dc52c72c5fd0e72b0c92d95":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-07-20 17:04:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/847139/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/847139/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/847139/5"}}},"commit":{"parents":[{"commit":"00f453372ccd4faf99e9a0bf527762a131968ad0","subject":"Merge \"Immediate caching of an image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00f453372ccd4faf99e9a0bf527762a131968ad0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-22 09:13:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-07-20 17:04:02.000000000","tz":0},"subject":"DNM - Image cache tempest testing","message":"DNM - Image cache tempest testing\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/847193\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/825112\nChange-Id: I01e8550aef85d333a70392be83a964267e632c2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3b8a54ccd6674bb1dc52c72c5fd0e72b0c92d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3b8a54ccd6674bb1dc52c72c5fd0e72b0c92d95"}]},"branch":"refs/heads/master","description":"Rebase"},"1ef06ef08ba177bd9ea93c25f96fa13aa6a2e9de":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-07-27 20:06:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/847139/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/847139/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/847139/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/847139/6"}}},"commit":{"parents":[{"commit":"00f453372ccd4faf99e9a0bf527762a131968ad0","subject":"Merge \"Immediate caching of an image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00f453372ccd4faf99e9a0bf527762a131968ad0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-22 09:13:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-07-27 20:06:55.000000000","tz":0},"subject":"Enable Image cache tempest testing","message":"Enable Image cache tempest testing\n\nThis patch will enable new cache APIs tempest tests in glance gate.\n\nChange-Id: I01e8550aef85d333a70392be83a964267e632c2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ef06ef08ba177bd9ea93c25f96fa13aa6a2e9de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ef06ef08ba177bd9ea93c25f96fa13aa6a2e9de"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
