)]}'
{"id":"openstack%2Fglance~847420","triplet_id":"openstack%2Fglance~master~I79e934f34f33953c821b6529a34e44dcf2a9688c","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-24 07:06:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I79e934f34f33953c821b6529a34e44dcf2a9688c","subject":"Remove options for oslo.db thread pool support","status":"MERGED","created":"2022-06-24 00:22:08.000000000","updated":"2022-06-24 07:07:47.000000000","submitted":"2022-06-24 07:06:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847420","meta_rev_id":"79a658fbb27b64a1a8551da70f6ca49a2a98fd49","_number":847420,"virtual_id_number":847420,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-24 07:06:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-06-24 05:36:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-06-24 05:36:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 01:45:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-24 05:36:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"5c609bedc68a52ed6b9d6ad2eeed7acd649542fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-24 00:22:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d68d7c6a0c2c61ef64bbd9ae81ab0b891ad42df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 01:45:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/be08901946eb4b5190b05af87f50b130 : SUCCESS in 50m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ad12324b5524d87a98e44ed4672e428 : SUCCESS in 1h 21m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a47ae42a61b34566a77654ab6a357a2a : SUCCESS in 43m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/020bc8b71fae414b9c873e0d44f66566 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97839c1ef2c9488eab711e92d602b0ec : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d928addcbaf46609495084b4571fc02 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba055e7482614a4a9bb6db70ad181794 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b1a4396b577494bbe20506e942e50a6 : SUCCESS in 6m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7644cc57df83457694cb0d2d3d940113 : SUCCESS in 10m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a90258702d5745c1863ad922bc4d04f8 : SUCCESS in 9m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b68b4076933541cd9c7f71105829eef9 : SUCCESS in 9m 47s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2e30dc18b4ad46289e42a5fee5a37315 : SUCCESS in 1h 08m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8ee016a66730444fa3fceecc01700ae6 : SUCCESS in 1h 03m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6c6d36d410064711a761b1f0c7741c3c : SUCCESS in 1h 04m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/45726272abd741638b58d123277ec8c9 : SUCCESS in 1h 12m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07346132ca814ad7b2df771c79b359bb : SUCCESS in 47m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/63fb415c137e4708b726e28599964ca4 : SUCCESS in 1h 15m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/672b19b6fd81447394be9502f1cd3d42 : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b15022cd33f14d595c5501b0d3ef0929ed92aef0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-24 05:36:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91e962fab7fd5c3d23f65f2d354a409ca213a710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 05:37:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7caa1a65c9a0cbc0b7bb270fa85c8d3b82b816c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 07:06:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2737215b2e2c49d58f10ce2786506c65 : SUCCESS in 56m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/664f067d81a74141b3157066968f0981 : SUCCESS in 1h 21m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/975310008a184af1a8a84d5c52920a9c : SUCCESS in 58m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e8a36ad3e84fddbe9fe67531ddf256 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09eee815076e455ba97517a54bc53715 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e62d38d498c42cc8d158e19fc0a7329 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/756c7681bb0b40ecaad2e3db459876f7 : SUCCESS in 6m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bd95106885bd4b4583a584f148cb0b9b : SUCCESS in 8m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6f85df7bb30747059c60283ab7f60916 : SUCCESS in 8m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/83816e1f757f44d6aab5c681b1fd3110 : SUCCESS in 1h 22m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ace5be4c0a3f4d0ea973f15d568b9010 : SUCCESS in 1h 22m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f2cdf3801a548e196bb464b377cedca : SUCCESS in 55m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ba90e6b009ec4b8bbd0dfa035acf6cc0 : SUCCESS in 1h 08m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b0bc4b6b90c644db8193db57ec7c61ec : SUCCESS in 33m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7f9f1aec395bbb56f8489f7585b78311ff853a06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 07:06:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"79a658fbb27b64a1a8551da70f6ca49a2a98fd49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 07:07:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c59c23a9ed34190b61859a4f0be994f : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08a0a738b35fd6dfa105068f0032bd5e84704909","revisions":{"08a0a738b35fd6dfa105068f0032bd5e84704909":{"kind":"REWORK","_number":1,"created":"2022-06-24 00:22:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/847420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/847420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/847420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/847420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/847420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/847420/1"}}},"commit":{"parents":[{"commit":"53f322f1d8c0a5c1e8c494f88f1f05c47b29729a","subject":"Bindep fixes for SUSE-like systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53f322f1d8c0a5c1e8c494f88f1f05c47b29729a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-24 00:18:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-24 00:20:41.000000000","tz":540},"subject":"Remove options for oslo.db thread pool support","message":"Remove options for oslo.db thread pool support\n\nThread pool support in oslo.db was already deprecated[1], and it is not\nactually used by Glance.\n\n[1] 74c6bf266e0b86ceb4726cad9268f134d423bae8\n\nChange-Id: I79e934f34f33953c821b6529a34e44dcf2a9688c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08a0a738b35fd6dfa105068f0032bd5e84704909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08a0a738b35fd6dfa105068f0032bd5e84704909"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
