)]}'
{"id":"openstack%2Fglance~848001","triplet_id":"openstack%2Fglance~stable%2Fyoga~I9d1c25638584fe346e3937d0536413d548d46e8e","project":"openstack/glance","branch":"stable/yoga","topic":"backport/bug/1980049","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2022-07-05 01:12:53.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-07-05 01:12:53.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-07-05 01:12:53.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-07-04 23:40:04.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I9d1c25638584fe346e3937d0536413d548d46e8e","subject":"[APIImpact] Correct API response code for DELETE cache APIs","status":"MERGED","created":"2022-06-28 17:43:28.000000000","updated":"2022-07-05 01:14:08.000000000","submitted":"2022-07-05 01:12:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848001-backport/bug/1980049","cherry_pick_of_change":847877,"cherry_pick_of_patch_set":2,"meta_rev_id":"dfc288fc7ad60177a7ad04c45c7c7c1c29e5848c","_number":848001,"virtual_id_number":848001,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 01:12:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-06-30 04:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-07-04 23:40:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-07-04 23:40:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-28 19:24:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-30 05:00:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-06-30 05:00:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"97674646e38b10aa351b8df4336cf947d1702fe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-28 17:43:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c7b8ed04aefd039b5e3ebca45ba2803901b77d1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-28 17:44:55.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"285d1d310153d17e1b11a3ddc6389e259c1ab261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 19:24:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a70fd1013a794a339de366388c217cdb : SUCCESS in 47m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1a7a0839937348f08d67867d9d9a7fb0 : SUCCESS in 56m 53s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0879eb2e30c440ebac034a9406479f3c : SUCCESS in 1h 05m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba30b3b4e93b455c80dd281f8381c298 : SUCCESS in 46m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49cb2a573394670b429fa0d39aa39f9 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6897f0007ef4561a51a8a43fb2eb799 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/593eae87ed7c433da6049d196175b64b : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac181786e9d47f9a7c15c10d81ec7b1 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/082fb0d8f235480f998f22b56a1e072f : SUCCESS in 5m 46s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/dbe5bb6b64e7449f84866b245ddee4cd : SUCCESS in 11m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9b2a49e18934024ad8e1aaf53766a08 : SUCCESS in 8m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f5217e5a8c2f44358cb06d0bfbcecf8b : SUCCESS in 8m 27s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/77098ec20ddc4f52b145a8f31be49090 : SUCCESS in 4m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/97886dacc4f549c2825c88a2d9e0466c : SUCCESS in 1h 11m 11s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e18b26270f5640dab5dc39de8711a245 : SUCCESS in 1h 33m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bc78393465c9430f9db7ddb8af346b16 : SUCCESS in 1h 20m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/742092f86f00481994d81095466c54e8 : SUCCESS in 1h 34m 00s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/2b36bc70db154201a082eb76660263cb : RETRY_LIMIT in 19m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aeeda735762b4fb088701bed5a2f2b53 : SUCCESS in 40m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6089094fd20b43f1931c5b4fd8382e52 : SUCCESS in 1h 15m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/01bea55e744c43459092fc2d4c9d1927 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f0ac1a5e960d9b956d5c92063b4e9e56e29f2db3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-30 04:58:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b17e6c452296555ef2ebe8dcf56d5eeb57d24cb0","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-04 23:40:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d0df803198d7b4c4393231f3d49e29abfb5cf9f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 23:40:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"24119aa84c2278ef9821db81edf99dd5fa1b2939","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 01:12:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0c47356eef34fcc9d48eaab2fc1fd46\n\n- grenade https://zuul.opendev.org/t/openstack/build/ed6648f29077459797f35c50474990a0 : SUCCESS in 51m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de8a0ff8d5e94ee3a8a868c0f1d344db : SUCCESS in 1h 23m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc5bb01a26234b54bf3a5b64a647d58e : SUCCESS in 48m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c7f40121ba49ad8b8e536a2108e4f5 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cb43d866d4b49f481d7dce5039f8dc1 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/489c572bf3c8410b9cc2924cb83fc9c5 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b762fa697ab146ed8bc0ed4ab63b684a : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9896a51ccc1454ba10f91b3d4a4afc4 : SUCCESS in 4m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25a869fa79dd44d68a1595444f1d9114 : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b5d367b8b86d404ab1ecbe99a36cf584 : SUCCESS in 8m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/feeca156df9b43f88d6f3d240f023878 : SUCCESS in 1h 25m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/43dbfad2d525476faa2c776e3916b02a : SUCCESS in 1h 14m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d447dc66ca24b4a93fc5e061aad6739 : SUCCESS in 1h 01m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/28bf34e7536c4c5ba51e8227f57deacb : SUCCESS in 1h 27m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/64d3d11ca59c4eeeba21dfcc31e3d843 : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":2},{"id":"56445775c0b01bd06bb05192713f093092328f2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 01:12:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dfc288fc7ad60177a7ad04c45c7c7c1c29e5848c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 01:14:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/354fe44c30614f8cae2e25aedb34162b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb0dd36ff1646b4b89fe23586237371 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/727b3cbaecd74d339c782a8907baa26c : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea4ed6dfec2ade544ba30e3324e6c7def6a3eddd","revisions":{"a9b297630024a0c258dea5c0cb88a13ec4bed0ee":{"kind":"REWORK","_number":1,"created":"2022-06-28 17:43:28.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/01/848001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/848001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/848001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/848001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/848001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/848001/1"}}},"commit":{"parents":[{"commit":"1964876644c7086eff516832a19206d28891110d","subject":"Merge \"Bump Image API version to 2.16\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1964876644c7086eff516832a19206d28891110d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-28 05:40:15.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-28 17:43:28.000000000","tz":0},"subject":"[APIImpact] Correct API response code for DELETE cache APIs","message":"[APIImpact] Correct API response code for DELETE cache APIs\n\nDELETE /v2/cache/{image_id} and /v2/cache returns HTTP 200 response\ncode to user but as per proposal it should be HTTP 204. This change\nreturns HTTP 204 response to user.\n\nCloses-Bug: #1980049\nChange-Id: I9d1c25638584fe346e3937d0536413d548d46e8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a9b297630024a0c258dea5c0cb88a13ec4bed0ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a9b297630024a0c258dea5c0cb88a13ec4bed0ee"}]},"branch":"refs/heads/stable/yoga"},"ea4ed6dfec2ade544ba30e3324e6c7def6a3eddd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-28 17:44:55.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/848001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/848001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/848001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/848001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/848001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/848001/2"}}},"commit":{"parents":[{"commit":"1964876644c7086eff516832a19206d28891110d","subject":"Merge \"Bump Image API version to 2.16\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1964876644c7086eff516832a19206d28891110d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-28 05:40:15.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-06-28 17:43:30.000000000","tz":-420},"subject":"[APIImpact] Correct API response code for DELETE cache APIs","message":"[APIImpact] Correct API response code for DELETE cache APIs\n\nDELETE /v2/cache/{image_id} and /v2/cache returns HTTP 200 response\ncode to user but as per proposal it should be HTTP 204. This change\nreturns HTTP 204 response to user.\n\nCloses-Bug: #1980049\nChange-Id: I9d1c25638584fe346e3937d0536413d548d46e8e\n(cherry picked from commit a71f7ba44e57a7945addb82afd2a19962d5c04cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea4ed6dfec2ade544ba30e3324e6c7def6a3eddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea4ed6dfec2ade544ba30e3324e6c7def6a3eddd"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
