)]}'
{"id":"openstack%2Fglance~848994","triplet_id":"openstack%2Fglance~stable%2Fxena~I09fc9164a08f42507d2aec44c5b382a72f232571","project":"openstack/glance","branch":"stable/xena","topic":"fix-rs-type-404-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2022-07-09 02:39:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09fc9164a08f42507d2aec44c5b382a72f232571","subject":"Fix failing namespace list delete race","status":"MERGED","created":"2022-07-08 07:30:16.000000000","updated":"2022-07-09 02:41:35.000000000","submitted":"2022-07-09 02:39:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848994-fix-rs-type-404-stable/yoga-stable/xena","cherry_pick_of_change":841935,"cherry_pick_of_patch_set":1,"meta_rev_id":"9f0a18e0d57182a724ffc35bb37e668d5fb51275","_number":848994,"virtual_id_number":848994,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-09 02:39:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-07-08 09:03:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-07-08 09:03:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-08 07:30:16.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2022-07-08 08:43:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-08 09:03:35.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"684d27e09bd27c4d799707091673152fe7193568","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-07-08 07:30:16.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"586d7b547f2718879b1f71de12935eec06f46e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 08:43:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbdba63d516540a2adcbe6e434eefac5\n\n- grenade https://zuul.opendev.org/t/openstack/build/d858a25cf0d74464a7b9a685916e33c3 : SUCCESS in 54m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ef919c2c12347f7bfcf9adc5d8c7c0c : SUCCESS in 1h 10m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6593587a51654ff6836f56702a59308f : SUCCESS in 37m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5195fc3a15f7463491a86bb71a24a4fe : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f9b87bd1c7e49f792bef43ad66a3ac4 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe10b449e3634ecdb8f0b8836d46e11b : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a934f96f9697409cbbcb16c9874b8431 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/295bae36f1964d51b5133ba030362d0d : SUCCESS in 7m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b20ed523b601433489fdb7ab5173740d : SUCCESS in 9m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/63870934610e4e119caecec20ffce129 : SUCCESS in 9m 33s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/fe47bf8b1fe142f286363d31d0e3ad2d : SUCCESS in 9m 40s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0b7d4595b9bc40c3900c269d11dad012 : SUCCESS in 4m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f1bc35d262b64ab1a9eab5b87c85173d : SUCCESS in 1h 11m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9e1ed429b45143288cedf61fa5b02b64 : SUCCESS in 56m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/220ae919c0d14f388c88771ca76636af : SUCCESS in 1h 00m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/31afb0429126424da2918fdbdbd6db99 : SUCCESS in 1h 08m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bab6aaeebbf14e8ea937a69015dc4e19 : SUCCESS in 48m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ca187438df314757a3458ee571018568 : SUCCESS in 1h 00m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ccf2bbe8ff0448f3ad8f006d89e28e23 : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c69b558bf516633ba50ef5ee11ffd2f7ddec743a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-08 09:03:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb54d9bbf3416b9853f851d208a1510d78c30914","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 09:04:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c79125b4e92faa345334aef53253e8274955800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 10:21:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcaf94ec23a14671813cbb69ea519dc4\n\n- grenade https://zuul.opendev.org/t/openstack/build/f53e124f74a5473eb5a245523e057a48 : SUCCESS in 54m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/61cb47da67cc482a85f05545e586f114 : SUCCESS in 1h 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d501ab54c06342e1bed9e61017704f04 : SUCCESS in 42m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d3a25982d064553af19af8cf2abf017 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f33e3f0d75034a288d2fc3feafb7ecf7 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec30b08f38f348e08ba98a48875b0bb7 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/737f45e4a3a8433fb40717e157331d8d : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/824d65914fd04aed99c4259a57193802 : TIMED_OUT in 30m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c3657a44d9ea4617bbfc25491d7798e6 : SUCCESS in 7m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ed71d742928a45d7b752996a01f5c668 : SUCCESS in 1h 10m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3917f30cbf6a436ba959decd3815a0e1 : SUCCESS in 1h 10m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2364ee9c07c41279b7a59f8bb768690 : SUCCESS in 39m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6fba3677cbd249efb0d240aa8ccadc5f : SUCCESS in 1h 12m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/824a3061155d4a7185ce88acffbf1faf : SUCCESS in 27m 47s","accounts_in_message":[],"_revision_number":1},{"id":"6e76fca0b80a37666dca9eae3f4771d3036e0194","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-08 11:26:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c63f39e7f1545eff4344767892772255958d4935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 12:54:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc1b9cce98e24f16b28197f914aa00d5\n\n- grenade https://zuul.opendev.org/t/openstack/build/93fd7e983ba3415fb22852256dbe4897 : SUCCESS in 1h 07m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5831568c698b4f8eba2004548d509e1b : SUCCESS in 1h 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92d384abc00d4a418951e91137ef55bc : SUCCESS in 48m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f207c7ad29a64f7a9857c7a15ef5c12d : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96cf8cc2bb1e4d69bdaf0341f43869b8 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd6a7e7b3d9a4dcd8356f27e3c6a7575 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/085e348c2d574319bea2738de0fce660 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c4ac69c71364132b353d49b20fec333 : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/06ad201d4c0d4c78a779998eb1d1a746 : TIMED_OUT in 30m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9b899a5aa1e24c4495213ac55a633e09 : SUCCESS in 7m 42s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/7255d09bbc344924814d7b9019796709 : SUCCESS in 7m 47s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8b27ce0233f440e5974797ca98929915 : SUCCESS in 3m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/58d806eb4adf4c98992609d60f9d9841 : SUCCESS in 1h 05m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/899731266c9e43d5902ea86eb846a6f0 : SUCCESS in 1h 12m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cc32cb227b2542b3bfe3cbadde60fcf8 : SUCCESS in 1h 16m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2bd14f32c76749048ed107f9da039ee1 : SUCCESS in 1h 17m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/285c42e1e1534c11a2e9d1bf2dc160cf : SUCCESS in 50m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b38b24ceba5546debdc958aa3e502872 : SUCCESS in 1h 22m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/53b9d998a64648ccb5ee5397bdbec4f9 : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0e93ebf62d54a6279b81fb82793212bbaa729f98","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-08 15:51:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a98cca9768d9e168f513d4935f24b1d66ad2e123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 17:23:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f597f669c34b09b6d942e192055647\n\n- grenade https://zuul.opendev.org/t/openstack/build/c49363256c4147e59aa7e4334c117450 : SUCCESS in 1h 03m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85bb6cefda234095a7bca23ba9107d4d : SUCCESS in 1h 17m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bbafa41cee54461bfc434b1f9d786a1 : SUCCESS in 56m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95fc7313194a4229b9f2c97b73866fc9 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d81833d530f548fbbb7d5dda269379d0 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30ead8b5039f43a7a23ef84c7259c4f9 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6633897e86e940d39a172042d0120596 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a4eef260174483bac621ac22c856c4 : SUCCESS in 7m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87be868594164a7d98d1c924ef142348 : SUCCESS in 8m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/34704b4ddda94b81b99e746d48e3de2e : SUCCESS in 8m 01s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/cea5f041efc84d2a8f16e8777781067a : SUCCESS in 8m 16s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6ff4853cf27641d0a791be8f0f6561bc : SUCCESS in 4m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c0c1a36f94b242f39def9a18cd7d09e1 : SUCCESS in 1h 13m 54s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b77cd631e152405296e0ae0d8f1e38dd : SUCCESS in 1h 25m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3923db3f93224b41979bffc7bdda2ec9 : SUCCESS in 1h 23m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/97e88076ddd144a982c5bbb67d08a391 : SUCCESS in 1h 27m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb78fa7206c44e20b857810637607729 : SUCCESS in 55m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/60ce94776a84429c8337ebfc83702989 : SUCCESS in 1h 20m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/afa4fe23602e4179921e935455562fa0 : SUCCESS in 37m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a6d9bbf59fba4a239ff87bb016a59952d19a09a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 17:23:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2884f8d2e3dfe03260aa4e735ea7c2dc3483f1b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 18:57:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/713cf4624c9c4dc2a9c34ba8e06e5d70\n\n- grenade https://zuul.opendev.org/t/openstack/build/53ae8b2f1f274ad5b679a8c1bba73f58 : SUCCESS in 1h 09m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6594960304dc498a92b59adf185028a9 : SUCCESS in 1h 27m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68f1e10459b649bea106b4c077fdf5e1 : SUCCESS in 48m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8953a6ffb9234c7c9804037b3ed299b9 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/237e9ad3623e4dacbe5d1fc9ec1c7891 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/191e25383aa14a52b7092d9876c2f37c : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee606989523640dd8659ab52c64503fe : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0d24ac492684f7b83aa3cb5a43e6da8 : TIMED_OUT in 30m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d6250764186e45aa8b333df8c3efee4e : SUCCESS in 7m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c86541bd6e6d44eda62dfdd4a2b1353a : SUCCESS in 1h 20m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/52041fb9b3cc4887953fee31f69e825d : SUCCESS in 1h 15m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bb22992b8bc4467ab40b53814b9611a : SUCCESS in 52m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/da5081661a8a45c283ea39ed9c31d954 : SUCCESS in 1h 19m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/34f9395bb8324c5caedcc999885653d6 : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":1},{"id":"27c6d40eee685220eec0b8260643df2227e61a17","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-08 19:17:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b948d0f53a5d8853ec0b7403ec67e3094ea5b279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 20:37:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8df6bb0ffa6440b81bee1548c679e5f\n\n- grenade https://zuul.opendev.org/t/openstack/build/2bef2b2d266c4b51b8981c2a315fa4d2 : SUCCESS in 57m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/585967c637c34d80ba7ba00390ed5922 : SUCCESS in 1h 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e1c7f9b5a4445b49b06a05bf63cf372 : SUCCESS in 40m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540f09e32580418fba47132ceaf96265 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a30c976ab654a84899a7c124c67bd32 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2520e5d5879b4ce2abc861d2e54d76f4 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/087cedf653764e19a53ac26681a80c86 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120820368fd5402e9043b02e0672d344 : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/21e2d37a500049a1a8ab720b4eaf74fa : SUCCESS in 9m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4f804a8dba9a4c35bb0d88068bf1f998 : SUCCESS in 6m 30s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2095af6946f54a6cb58814796378e687 : SUCCESS in 8m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e847381904924bb69656d845d7aacbcb : SUCCESS in 5m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1fbf7ca3eba141cfa195061fee574c4f : SUCCESS in 1h 13m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7fa06ee83d5e494281581514bac4f7ca : SUCCESS in 1h 11m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/430d8410808848f985a58a86cc3f37ed : SUCCESS in 51m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3b79d8f0c8154fa28eb06331859c24e2 : SUCCESS in 1h 18m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d633ec12bd454531bf07f413de41896f : SUCCESS in 48m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/221a7b432dd2452bb1de71d5083c201e : SUCCESS in 1h 04m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d757c14f1d994f2d8e3047ae891f1315 : SUCCESS in 27m 51s","accounts_in_message":[],"_revision_number":1},{"id":"cca6f5b870905c21cee291d5ef38b1ae8981e85f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 20:37:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed44213aa21748c4094e8013ae3696deff7dc495","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 22:10:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a0f1116e5bb4acabe05bcb902796669\n\n- grenade https://zuul.opendev.org/t/openstack/build/49ba928965474937b0d20243703862c9 : SUCCESS in 1h 19m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84072a5c6a814eeba3c60f2ac67145e6 : SUCCESS in 1h 28m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1d69442c0804ccc8d291bc495643502 : SUCCESS in 41m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bce8b2627bf14d379cffbcef89b0f607 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45bd65b36f694204acf6843f16f75fe7 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b8bba5fe3e246349b1f794480ab8d1d : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a405c7675b4d659e7278c99d4d167f : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63d77292d1fd472497aeeaee02b7a703 : SUCCESS in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/21f3f337863a4fa887a57d72cfa756ce : FAILURE in 9m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8209e49ac0eb4da1b960ced80d0a9f42 : SUCCESS in 1h 12m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a3f73f90d5b84abaab2ff40dc160425f : SUCCESS in 1h 06m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/beb427a6fe51404e99135a9ccd7b71d2 : SUCCESS in 43m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/52101c3e523848d6a7470b32bf78825e : SUCCESS in 1h 04m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/91385fbded5649e1a013d153f9fe90eb : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3ee7843e59e5030e9434cbf5eea4fbbe51fc9e7e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-08 23:48:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e3ab589b0bf76e0a511a4c5f19a536f268f8522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 01:30:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/619a301a83df4538ab050ff5e423139c\n\n- grenade https://zuul.opendev.org/t/openstack/build/ec89d12c0e5647d2a15ca9f7e9d87b78 : SUCCESS in 1h 01m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c1832f6991344a0bac9ebfaf99a85de2 : SUCCESS in 1h 19m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b370203c6c042729ec0f6508683f144 : SUCCESS in 48m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f961c3ef326d4b8aaa50dd0258c822f8 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/247821561de14df19725edfe4a5f7001 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfaadca8f5734d168106ac525213f823 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9ed9e2f5ae64d5a9175374ab71fcbde : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57471e180dff4d3997ea07c165d0b601 : SUCCESS in 6m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8126dc1dbdae49d88628b8935fce3a66 : SUCCESS in 7m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eb3efff323b348708e86de05350bde08 : SUCCESS in 7m 42s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/dda8c61f2187495ebb3f30fe110a346a : SUCCESS in 7m 42s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4db387f442314f76b858b60ce618e9b2 : SUCCESS in 4m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5be25ef541e0440387a2298566e1af94 : SUCCESS in 1h 11m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/72ba55837cd743209469f8ef1fc30dcb : SUCCESS in 1h 35m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9068635aebc94aab85a2ca57fc624f0d : SUCCESS in 1h 12m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c4ce60538e6e458a85fce4927ce5cece : SUCCESS in 1h 15m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dd5866a211744349875cfc7273de51d : SUCCESS in 50m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/753ec3464f8a4e21bd94e83bf358402f : SUCCESS in 1h 09m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/db4526697f304e178f9abd2a5fd93e57 : SUCCESS in 35m 35s","accounts_in_message":[],"_revision_number":1},{"id":"af027a5c29ae20fded2149b55f8c0ba38f19c090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 01:31:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"534c01313820b28d10786d9b6b8a1eccba5c0e89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 02:39:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cd7323fd7e148c9aa673c74e85314cf\n\n- grenade https://zuul.opendev.org/t/openstack/build/03e9588478cd49cdaadb71c6ad69a120 : SUCCESS in 40m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2170d3ea4fe842efb5c049ad1d36dd8b : SUCCESS in 52m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acd5b47174fd411fa557026b64462cfe : SUCCESS in 42m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9ee1825c8f94e428bfe035a5b5be52c : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b9df7346f304f1c9128153ecd06994c : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df7451e23a094d69ae35d89eb9a914ae : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/543dc9336f3f4b4ea63944cb8f26b743 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bef309c1ee8248bf8eb414b76d879749 : SUCCESS in 9m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/24386a8d440848109e987ebd635ae756 : SUCCESS in 7m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2c3c125bf16e4f17a9e71a2ba87021fc : SUCCESS in 1h 04m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/11fdef0a43ae4c49bb7b8b1f7993911a : SUCCESS in 1h 00m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a90d1192ab5434ba37a28eace19f2be : SUCCESS in 35m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2f76f18c3db747eca1f26bd68dfb8df0 : SUCCESS in 1h 07m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0f0cb11090cc40df80a048605647c38c : SUCCESS in 22m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c51e90cbf0c513a6a8000466149eda3ee1044eda","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 02:39:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9f0a18e0d57182a724ffc35bb37e668d5fb51275","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 02:41:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f49cd086b7ca476bb8b8047a57d72c10\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a73e2e959804540b98c3dfb10369582 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc9a53d257babf0ef7f0bec12ffff46fc570910b","revisions":{"fc9a53d257babf0ef7f0bec12ffff46fc570910b":{"kind":"REWORK","_number":1,"created":"2022-07-08 07:30:16.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/94/848994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/94/848994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/94/848994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/94/848994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/94/848994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/94/848994/1"}}},"commit":{"parents":[{"commit":"316ce9659acdec0bc45f8a33825ae3b076199b24","subject":"Fix set_property_atomic() boolean type casting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/316ce9659acdec0bc45f8a33825ae3b076199b24"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-05-16 18:58:05.000000000","tz":-420},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-07-08 07:30:16.000000000","tz":0},"subject":"Fix failing namespace list delete race","message":"Fix failing namespace list delete race\n\nIf a namespace is deleted by another client while we are doing a\nnamespace list operation, we will fail the list with NotFound if we\ntry to pull the resource_type_associations list. The latter re-queries\nthe DB for the namespace and will raise NotFound to us. This is\nespecially bad because the namespace being deleted need not even\nbelong to the caller of the list, as is the case in a tempest run.\n\nThis makes us catch the failure and continue the operation, reporting\nno associations so that the client gets a consistent view and no\nerror.\n\nCloses-Bug: #1973631\nChange-Id: I09fc9164a08f42507d2aec44c5b382a72f232571\n(cherry picked from commit 35e6c57bc90365796f45ada9271e532068557b8a)\n(cherry picked from commit cc98dbef31be30372b59b2d18b78694ce0879aa0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc9a53d257babf0ef7f0bec12ffff46fc570910b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc9a53d257babf0ef7f0bec12ffff46fc570910b"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
