)]}'
{"id":"openstack%2Fglance~849121","triplet_id":"openstack%2Fglance~master~Icbb524307b669a6ab5e5f9b7b17f315450ef6a37","project":"openstack/glance","branch":"master","topic":"remove-misleading-note","attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-07-08 15:00:00.000000000","reason":"Erno Kuvaja replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Icbb524307b669a6ab5e5f9b7b17f315450ef6a37","subject":"CONF.paste_deploy.config_file: Remove misleading note","status":"NEW","created":"2022-07-08 14:54:00.000000000","updated":"2022-07-08 16:35:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1dd917bbdbfa9448009ddc1b9462628a81a7c4f9","_number":849121,"virtual_id_number":849121,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-08 16:35:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-1,"date":"2022-07-08 15:00:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-08 15:00:00.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-07-08 16:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"230fc950e2d6a05646d7cdc7ff03b7fb91336ee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-08 14:54:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"251e57d55978b157154c51c48944a144023c6a86","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-07-08 15:00:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1dd917bbdbfa9448009ddc1b9462628a81a7c4f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 16:35:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce668fc4959746ce89fa2f3958cdb2a3\n\n- grenade https://zuul.opendev.org/t/openstack/build/91625eddc02e400d9a2bbbe02e3ea3ee : SUCCESS in 1h 01m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5f28dd1d72b429097865fd63416c7d7 : SUCCESS in 1h 28m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c9a4c2411314c6c864dd7d6aab55e82 : SUCCESS in 1h 05m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd1b918b481d4b2f827fa86b61784abe : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01546c34f77b4e639ad9f8675484e3d8 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93a1fb83e01d412099d4edf1f3df367d : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b38befab71b419385f0a7b1b5bc56ae : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ea688db039442a98d97853a9d47f0c : SUCCESS in 6m 42s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/283cdbcad7dc4aa6acd96a80aea5c26a : SUCCESS in 12m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/75daadad5bda4cd09da90d4a52f01002 : SUCCESS in 9m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/810142bd850a4de2ad4ee99aea7a5300 : SUCCESS in 9m 31s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3c3dbd87c6d64fc7b6ae38d86eb78b0e : SUCCESS in 4m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e9709ebd0291420291430ae2b2a2a2cf : FAILURE in 1h 23m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ceac84304cd24a10a24deb2441185a4b : SUCCESS in 1h 15m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bb68f1ed3cd44667b6e88b15fca3159d : SUCCESS in 1h 19m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/84f1952b6ddf4f1abb4e03c2a197fe9b : SUCCESS in 1h 34m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8bf088c968af4446baf2f3f37d3bdc28 : SUCCESS in 1h 12m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/37c7dcda83eb4dfa8d746e8b740254f0 : SUCCESS in 59m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f3ca5e5310704230ab7c3027605c9dc2 : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ebeb62d754ba46c9b75170d9eb6872308695bf8","revisions":{"7ebeb62d754ba46c9b75170d9eb6872308695bf8":{"kind":"REWORK","_number":1,"created":"2022-07-08 14:54:00.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/21/849121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/21/849121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/849121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/849121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/849121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/21/849121/1"}}},"commit":{"parents":[{"commit":"fcdbf8a4a4b28ad63efaea127e51f72d4b578118","subject":"Merge \"Doc: fix URL in API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcdbf8a4a4b28ad63efaea127e51f72d4b578118"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-08 14:47:07.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-08 14:47:07.000000000","tz":120},"subject":"CONF.paste_deploy.config_file: Remove misleading note","message":"CONF.paste_deploy.config_file: Remove misleading note\n\nCONF.paste_deploy.config_file can be an absolute path. RDO has been\ndoing this forever, and our tests use an absolute path (see\ntest_load_paste_app_with_paste_config_file for example).\n\nChange-Id: Icbb524307b669a6ab5e5f9b7b17f315450ef6a37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ebeb62d754ba46c9b75170d9eb6872308695bf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ebeb62d754ba46c9b75170d9eb6872308695bf8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
