)]}'
{"id":"openstack%2Fglance~849725","triplet_id":"openstack%2Fglance~master~I94ee03a00dd01726de15a29424183faabdeba844","project":"openstack/glance","branch":"master","topic":"fix-E741","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-02-16 11:53:32.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-02-16 11:53:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I94ee03a00dd01726de15a29424183faabdeba844","subject":"Fix E741 issues","status":"MERGED","created":"2022-07-13 18:30:03.000000000","updated":"2023-02-16 11:54:59.000000000","submitted":"2023-02-16 11:53:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"849725-fix-E741","meta_rev_id":"cd7c90a6a8a270849696bca3bffe56c85a765a66","_number":849725,"virtual_id_number":849725,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 11:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2023-02-09 14:15:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-02-09 14:15:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-13 20:06:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-14 05:06:24.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-08-11 21:05:26.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-09-27 09:41:47.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"9f2da000ad9cc2c88b0be729ccc959cef718e5f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-13 18:30:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"913dc53fda293e56d708f4aea487b4ccfec96acf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 20:06:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64673ea3d8284bd3a1dacfe4dda7ca70\n\n- grenade https://zuul.opendev.org/t/openstack/build/679ae1711f32481c80d6d17932c803d7 : SUCCESS in 50m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c159982e3c7f476cb101d9e81987c485 : SUCCESS in 1h 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0414b128ac30402f81ee94d8061140ea : SUCCESS in 44m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb232beab83d4700bc1dcfdf54a98ad1 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c8f804d762e45ca942812c1b898b0db : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f520668b463e4e25af334c635d395e40 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19bed203552346929e78da345851fb39 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69bcae5bea8c4ec382f8f36ca1381ef0 : SUCCESS in 7m 23s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/81f44b3edbd64bd78e0f61ffdb6daef9 : SUCCESS in 11m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/132d9a1f1e844de99c93efb94690dbeb : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e82c38a39ca54f0e9d8e5342d086677f : SUCCESS in 9m 11s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e96637ef4c3849a9afdf04174acff72f : SUCCESS in 5m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5e668204001e4b41add922c1ed89d506 : SUCCESS in 1h 20m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1d8df0a2401c48b6ae74a493ed966314 : SUCCESS in 1h 21m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/68d6dd93e47e411f97d6913686ac4749 : SUCCESS in 1h 14m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5e7ac9452ae94c1bb716d71bdbb26a7d : SUCCESS in 1h 25m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c5f2693ab774ba88607ee3403cfdd1c : SUCCESS in 53m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0fdd02eb808a476c9fe3628b9cc1e87e : SUCCESS in 1h 04m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb3b84dba90a41b79c12da56b14d429b : SUCCESS in 33m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4b42314047620ac35ad170b01f044c699d78be54","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-14 05:06:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eef4af81fb3567955bb39483c4aa251d46de2b27","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-11 21:05:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ade1afb296006f620070f05088bee1b8aa2f9fe1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-09-14 23:32:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7dbd22a0014d77d2cd1944c29f13790474b94af4","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-09-27 09:41:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2248fd566a2e388969599f24d7913df2aaa11d40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-01-16 11:16:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d65be9453962cdec37b0decc2e1271e5b404ddd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 13:06:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec40d6f088ca4f0ebca5ab0f33c947d5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6832f7aa78324dcfaf7f93648be8e17a : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/292352f7c13649e181d66b19367ea725 : SUCCESS in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2ef289304b2c436084f898e35c016e7013a3242e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 14:08:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e7ee5bd92f8454f916661fb3717581f\n\n- grenade https://zuul.opendev.org/t/openstack/build/0aa9e24666784e7b98439f0924f6b379 : SUCCESS in 1h 05m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7803c4cee7424d99b9168eef01daa3d7 : SUCCESS in 1h 01m 57s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cce187623ee84b3eba4f3cd0de4d93ba : SUCCESS in 1h 18m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfde79984f8e4513b3f6d579f452957e : SUCCESS in 59m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db844ef0a919456eb22889c5008c3e4f : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cb5d6a535a0496ba44a7594fed19ede : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8a7419dc1554d7090b507fa1c5fba51 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d613399a8f0e474ababca6ff8dfef861 : SUCCESS in 5m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/936fe6fdbc644c7885a79ab5b0dd4440 : SUCCESS in 12m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27fa7dbf7f8c4df695ae742f5061ec75 : SUCCESS in 9m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c3aee45c894144908bd836733e7d738e : SUCCESS in 10m 53s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ec16b9f2bd744c269045eb804a1ed93a : FAILURE in 4m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dd350b5c3d36417b894954c2c16829d3 : SUCCESS in 1h 34m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/462f5e02416e4f188977abe36aa22b0e : SUCCESS in 1h 31m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e8229c1254f5472080582178ae4650a8 : SUCCESS in 1h 24m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e9cc4d6149c34881961e2487018bd854 : SUCCESS in 1h 22m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1babc529b1c4a29b5f657146b53691d : SUCCESS in 31m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c624448a40264537b43c1ac61b5dd7a5 : SUCCESS in 1h 23m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de8d576556504a3d8cd820d578a2f2bd : SUCCESS in 42m 21s","accounts_in_message":[],"_revision_number":2},{"id":"9db4232f5bf16c0a453c90e8422e700dd64e16eb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-01-16 14:11:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0df65a8497ecbc777556f68d847952ea1348dff9","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-01-23 10:34:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"30206458dfd54067304c1e5c2eb1cd0fd3632309","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 10:47:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d4d82014c1c4944b019cc2f46121485\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3de1f774ac7f44289d74a0f486730af5 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cc1c707ca5bf42d3b54bc75793be0fc7 : SUCCESS in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"113459467e05ddd873e3e3206dc4ed416fc6557e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 12:16:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c4e2278ac3e4cfb8e6b2b04cf01f32a\n\n- grenade https://zuul.opendev.org/t/openstack/build/7733ff8672354c15b4e875ec94d42402 : SUCCESS in 59m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/eff2c9c26baa48af835133f79baa423f : SUCCESS in 59m 26s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af7a026f58584856b1f0baeaa637f947 : SUCCESS in 1h 26m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b753338e134549c88843091c85058970 : SUCCESS in 1h 00m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9f6c99ec8f443ab967e5a8b2e9df00d : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a330f29507e4c0e99154e94ff11e7e4 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf1254d7170646ff82b1f0b969a20c70 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/422810efb2d24cf59f6ae1409e6f6654 : SUCCESS in 6m 36s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d08dd33e915543339b585400096ca2a8 : SUCCESS in 11m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8b6a895444984999aecc161d35975de9 : SUCCESS in 8m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2cd186e52241468cae2f8df0219834ef : SUCCESS in 8m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/45a12ff14474488f9ae171101e18d004 : SUCCESS in 1h 09m 34s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/85620858f5d44df088554b043cb9904f : SUCCESS in 1h 36m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/41de5defec8d4adfacf081b437a86e75 : SUCCESS in 1h 31m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/24587d612eb64669bf8731b064cf67e1 : SUCCESS in 1h 33m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6070ef99f9ae456b8ac32c823dbeb2ea : SUCCESS in 47m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa5ee9225b8c461f9b7fcba301534204 : SUCCESS in 1h 19m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4d4fdec84d3b4c6785253994183dc7d8 : SUCCESS in 26m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4aff9399618d723c291fe268d852ab4d170c42fb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-09 14:15:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29378173271a71b872e1a37b6fccd55c34c16f70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 14:16:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"219e1df453529b2b263b5ae3345a944935eb29b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 16:40:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36e82881c808411dad2a30810fa6125c\n\n- grenade https://zuul.opendev.org/t/openstack/build/921ccc4ffd8040229de4b7dd29a68896 : SUCCESS in 55m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e4c9bfd66d1494895fffdc729b584b4 : SUCCESS in 1h 31m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6314fb7fff2c4ee88716d7ca27f0b46e : SUCCESS in 56m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f3363e959504329b8c8c72f31600859 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a87f0130e8d42d4bf99abc96819adeb : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7dd65443f5d641e7bb2f19167771f761 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f08bf05a210944a3b81542c67204ee05 : SUCCESS in 6m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d968cbf14bcc4d06ba898a121c1004fc : SUCCESS in 9m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/12ee40ea17ce489dbc46e2fbd03a502c : SUCCESS in 9m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/169e2c0aafc6410788e9dae9d9df1900 : FAILURE in 1h 59m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c3bd1dd62aae4ccf8cd8eed5d8fdaf33 : SUCCESS in 1h 56m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a82a5083585b4918b7a38e7c3a9222c9 : SUCCESS in 1h 00m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c621b189f2b845538a1b6507c029fdb8 : SUCCESS in 50m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ad3f716d3be4717909789c99a25b65a : FAILURE in 1h 05m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fa9cc82d54c1415282a256429f3829b2 : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c5b00e6e227e06b993671837389720c2922d9ff4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-10 14:04:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7b0b1df2886c460b726fbd645f9a96bcf4f28c63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 14:11:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5c7cd5f98e34ddda1a171d53829994a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/387900270a6a4d52864023e2bd03dde8 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5ea35d7384024512bf9f7f3ae5015028 : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"100c3a7dc72c0c31ed01b82b77170466bd7ae587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 15:55:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4814967be77433292cb925bac982a30\n\n- grenade https://zuul.opendev.org/t/openstack/build/09d5dbcb57e04edaa07d7a1a1542a86a : SUCCESS in 1h 03m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9effdfa5c663417abcbf81f25f810e7a : SUCCESS in 1h 10m 03s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cdd1f8e684e44b4f9aede09bb63780c5 : SUCCESS in 1h 46m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5df8ebcf03434208b4e81754e5e78732 : SUCCESS in 47m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b5756716b5840c2b021fc7c9ee36259 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abef927bde794dc584f94da0d9e4d810 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8887e6856f1447ae947c390684553ed9 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a7ccea4a89a4e45aead3eb3df26ec60 : SUCCESS in 6m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0c50dd8402534e30a25a4505466c2d2d : RETRY_LIMIT in 6m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac1fa81b259f45de88ae409d9c150be6 : SUCCESS in 9m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e1d26d589e80432b9afb32bdfc204612 : SUCCESS in 10m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1de4fd84fbd34654b1d5c80192aa6f23 : SUCCESS in 1h 41m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/156d1814de7349769ed930668c7467d2 : SUCCESS in 1h 33m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0d0843aa9ffe4efc89cc01bc7fd92874 : SUCCESS in 1h 17m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c14d38cac12f4249960c7384d784cf37 : SUCCESS in 1h 33m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f573b63a04b40dd8b449ea18cc78277 : SUCCESS in 39m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a7a0956983bb4371926153615a28da02 : FAILURE in 59m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/06f0d6b2e1824d1fb2c4caa1caf090b6 : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":2},{"id":"32268fba6250bbb621512ef9a99199c8f7577290","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-14 13:00:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5b21bd5f0260378ba82102c03ff52502373001a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 13:18:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fd52f068e3c45d29e5243ac0dea6d47\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd8cd810d7564fc6b8d63846decdc658 : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/77d2749a4699437490942a862122eaf2 : SUCCESS in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46bf2d0288f34502d50438bf2a36c17b0b7c78b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 15:51:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a45d3bbd0b9402197f4b712a6d84cbd\n\n- grenade https://zuul.opendev.org/t/openstack/build/12273c31900147fe874185fa7cd609b1 : SUCCESS in 1h 02m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8997021b843046a1aae03d6da1468b43 : SUCCESS in 54m 01s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd863f6e2e534feba67885e514cbb2f0 : TIMED_OUT in 2h 04m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/078537623f0c4b2ea5a4e783348da5d0 : SUCCESS in 59m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd052aa0b34c464e85b16cb0b71c39c1 : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3839c444e91f4d9e93ac4ec533155a18 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92ba49ff44884570b570875a8ed0376c : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f4d526d711946cbb6d76efbbaa2989a : SUCCESS in 6m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/37dc25d94e744079a121935efb57c3bb : SUCCESS in 13m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9d6a75f4470f4d5ab5684ec37c2beb3b : SUCCESS in 7m 59s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fa9ae2571dc14d28bcc246fc24f1d002 : SUCCESS in 12m 55s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9b62d74dc33c451d9466b5f14ab923a8 : TIMED_OUT in 2h 06m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f9a90e57f99f47ef8c3a356d34eb3b81 : SUCCESS in 1h 22m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1c71b34a0e9442e4b2a13cbfaf1ef2f8 : SUCCESS in 1h 19m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/95801de4720541b88fa6e2435cab821b : SUCCESS in 1h 52m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29f7c4f09e3f4829a19c25b910d412d9 : SUCCESS in 54m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bad8dc228e6a4f7d8cbc6d598780c453 : SUCCESS in 1h 10m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d5fb6a60dbea4c00884345643c676ae6 : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":2},{"id":"7350bf4eafa0edc5efc4b0a80882afcea241b166","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-14 18:27:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"869a9c62915c49696bd8458295251dde8cb175a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 18:42:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdaee855e1214ad4874c65cd04e64274\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ad538e86d4846b986b3c0fb91a2e1ec : SUCCESS in 10m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6383488e4f0542c2a1f63c400f5c27e9 : SUCCESS in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"caa12b0e45aeaa9f112c4d0a2f92b11c5fdf9ef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:23:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f740798ebc04c00b883d714533eb43b\n\n- grenade https://zuul.opendev.org/t/openstack/build/041047ea01ba4598bc5ab88088413c34 : SUCCESS in 56m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/90cb7a998d90432d8584a2da067df6ec : SUCCESS in 56m 36s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a2b0627f6204cb79da20103933f3744 : SUCCESS in 1h 26m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94519125c577474f9e1f3f306fcc8afc : SUCCESS in 54m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8af3d2d02a9b4fcb9c88ef5d2c10beb8 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcdd0d26b504447d9f6598304ed210bd : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa7aa52d5fc242a7b368e6ab0b337fa1 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6818c7a75f24c89a4bf70a4ae3d9289 : SUCCESS in 6m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fe225d3e2e0742869e35a2b00e7837af : SUCCESS in 13m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e06c90d84814e10be996ab48d247bda : SUCCESS in 8m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/76b36f358fe54cd19c5f43756287afc9 : SUCCESS in 15m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b9443959137145f79b6ef0e8668f802a : SUCCESS in 1h 15m 05s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a8e666d7e1df4f0481af88a5165c6a9a : SUCCESS in 1h 24m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/161c378d380f48ef9b0bd562e54cd5a0 : SUCCESS in 1h 35m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ca491f5493134f7bb56c8098af3deec3 : SUCCESS in 1h 28m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cb27d515633420f898d19064ee5ccd1 : SUCCESS in 50m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9128866615a94ea6b70e6d624d878282 : SUCCESS in 1h 07m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e954179fcbe46cda4d1db86000c189d : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":2},{"id":"4d4aa5cfbd19352686da52be7b46aa8353484d87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:25:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a6138a33d6de20fbbdc50f655446ec4be72564e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 03:11:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0352dd0491ef4f26bb8d36c014939401\n\n- grenade https://zuul.opendev.org/t/openstack/build/74848a96bf0342b2b756527a1f42a5cf : FAILURE in 1h 24m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0439aeedd8784f1d83c2cfcb4243e2ea : SUCCESS in 1h 33m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bd7a04d21c742a5bb385fcf0294c2a8 : SUCCESS in 46m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ef8bb8674ab42d9b8225ff80d250aa2 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3044285964ae41319301132300a1f0ca : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd3a19d634104e65aaa62a5c31f933db : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95df1c41421840da88e6e5b0b1bd3d60 : SUCCESS in 6m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0c374c8407d94e62acf57e0ef7ee100b : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/06c01db5378e49098f247ff7f2105648 : SUCCESS in 8m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3037b6a2b70446958d1eebd2ca5dd5e2 : SUCCESS in 1h 30m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dfbd5e70498f44ac98ba4181e227b940 : SUCCESS in 1h 27m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6d188b62d7b349f09883ed30bc8ba793 : SUCCESS in 48m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2024e359c87a47078a631907ed9c3c78 : SUCCESS in 41m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/14299585afdb4ca9b9785c63c6bc4272 : SUCCESS in 1h 04m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b61d0d5dba234b718a75f7a84f038740 : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c5c1dca063f1fb5acdc65ae37d814b6f09bba163","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-15 04:44:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be41bfef6c4532b7ddaf4e6e1659bec2aa06c36c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 05:00:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10aab7622f914c9f86240c041c4c6634\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b498dda7aafd4550b2c05b69b26214eb : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3ffa09ef834e473a8a6e47a338f32803 : SUCCESS in 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1db042daebfd1c1537218b7ade614d4317e86fa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 06:28:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5edc5ecd67014872ab99e63ea2d25760\n\n- grenade https://zuul.opendev.org/t/openstack/build/2ca6408b96b34b9995f98f6644762d9f : SUCCESS in 40m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a61823cd358440ba8a08b650e013127e : SUCCESS in 58m 59s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5001428ef12e438a9edd4ee74c6ab2a6 : SUCCESS in 1h 16m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f29cd44b36e345748a721dd53364025f : SUCCESS in 1h 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edcebd1f39b245ddaf1da6a6240299ad : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72f27f9a089e4e90adf29a3b92cf798b : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f705ade42e146c5a31fc4afc41ddf78 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c1a7a05541d460cb7d7c94470385be0 : SUCCESS in 6m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c6f5a274df1d4fd3a5d825de3a2f25b6 : SUCCESS in 13m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a4bb1edabf74aaaac6a5cb10c47dfac : SUCCESS in 8m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/583c9d4c50d743a7b99214b425f876e6 : SUCCESS in 8m 27s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/30fc5c3814a74a28855f15af82b4b64f : SUCCESS in 59m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/473071c82e9742caadf4fc8e40face01 : SUCCESS in 1h 04m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0f7f24176b4147f9822900828b1be063 : SUCCESS in 1h 23m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/08a623ea7e9944bcb8995b48385cfbb4 : SUCCESS in 1h 30m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2c7d9f8711b43afa5f6dd5af8dbbd5a : SUCCESS in 52m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/663a7adb9c074bf68b2118afec51d751 : FAILURE in 1h 04m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/19b4bd7e37a246a495aa570f9eb5bc51 : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":2},{"id":"8fe1249557525a178ba51d9fa42ae181eee2c1c5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-15 06:49:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d860fb447c03594ed2137fb0b9fc204310e717bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:00:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82fae31cb87744d1b7580064dc8fa038\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6873ea382de4720a350c6b44e195770 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f70eabba658249848cdd93aa83c0c691 : SUCCESS in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30742c4044f63154b7b03f829a973e00c10a1208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 08:30:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85a5f2d5d806496eb721a461234e683a\n\n- grenade https://zuul.opendev.org/t/openstack/build/44160112eefa4e8296c35628468436ec : SUCCESS in 48m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/04bd11d09e204bd791f4098ac3695790 : FAILURE in 1h 02m 25s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3c5bff63d824a60a7425e2660ddb61e : SUCCESS in 1h 14m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b25c6796f254e73a6360000962564dd : SUCCESS in 55m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4610634c64554b40b938b616383e592a : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/798e3460799d420fba376e4823705655 : SUCCESS in 7m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/666b4aa364b64428a85226ffb249a557 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70d7ac06eef742a3816cfa43198eb538 : SUCCESS in 6m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d9b10b875ca144468be6dc9d882bdc65 : SUCCESS in 15m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d289e4003b394a8795434c2ce2a63d13 : SUCCESS in 11m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ff81f7f51fcb45028b7c4e2a491eca32 : SUCCESS in 8m 35s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/061c25beb9e341cbbbb97de8b47f35ff : SUCCESS in 1h 29m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/274f04306b3e4ff9af06fc09a89734ae : SUCCESS in 1h 38m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/598a25e3c12348a68f4cc7228d0741e0 : SUCCESS in 1h 25m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/67ce71d2c8464212ad5320066996d5d7 : FAILURE in 1h 27m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50f97c4ca2a74754b9db605c10b1e100 : SUCCESS in 41m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5583f0b144a44206873a77be4b513c4e : FAILURE in 42m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b493a4ac7464dbb98f52b2bcbacbf1f : SUCCESS in 28m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b59a2ffec4d19b1909fb631d9d2b6afa0e33799c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-15 08:40:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5ebeafb9f71cad0933f277caf3efcf1cf095df6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 08:46:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab00c6defa7f41eb9abcad883187803a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fc83e6066df45dd94e77ccf64084289 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aaf69eaa224e40e8bef7c35e253ad18d : SUCCESS in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff7eae1a9ec7b03b2c962356ceb17ce5407292c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 10:22:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/805079475a424b758b0523892ef4050b\n\n- grenade https://zuul.opendev.org/t/openstack/build/14e38c8c966443c999c953e9864bb268 : SUCCESS in 1h 00m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b112cd70db324c9f92e7074528b485d5 : SUCCESS in 53m 25s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9730b99d73a74a7c8f40d53f73fcffe0 : SUCCESS in 1h 19m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15d5a2ab40c240779b2e0fda483d6656 : SUCCESS in 59m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6568d72dd40e409a90bbd66c460045f5 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/832b8975b7f640b090f4792fb76fc7db : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2365074f94464f209591edaa0c86960b : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8625e2c2c4194fa8b50adac97a28ab12 : SUCCESS in 5m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e22a783e35e04082a5500f162422a27f : SUCCESS in 13m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/afc43c06e58a4d4397f6ea0a16b16c72 : SUCCESS in 9m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e4d321ca52fc4b93ab38c95f1d920488 : SUCCESS in 10m 20s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d91d8b8056ec4f81bc2873fdf59c3ac9 : SUCCESS in 1h 10m 56s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5be7dae57e3b41d0a715a2ff0aeef10c : SUCCESS in 1h 15m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/262835e4232d44049b7356a99dc2fabf : SUCCESS in 1h 15m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5eb3e90e2bb84c7287c49f608c38c14b : SUCCESS in 1h 39m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef4ebd7eac33475f9679144d78d1c9c5 : SUCCESS in 55m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fc301d9832cf461f8cdca718576af844 : SUCCESS in 1h 14m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f133067a6cde42de81810a3ffce782d2 : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":2},{"id":"118cacfa5445ff0f1d94b5b29fff3cfa4e5b7714","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 10:22:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b99c5a99219d600ae3c33122c18e7c7e54939b10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 11:50:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/91e9fb0a1bb747fdb72ec23d5ef394a2\n\n- grenade https://zuul.opendev.org/t/openstack/build/f45e9e4ebfaf41a4b78db8b09e185151 : SUCCESS in 55m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d820cf71bb964192a4aae61dcb94790c : SUCCESS in 1h 23m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e016cf1ca3248a4b5e9e9768567db44 : SUCCESS in 54m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21b36eae525849cfbaca6188de81ef9e : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e56320f32d94174afa509d9d03c7392 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/142f75bc33744654be770379d8587860 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031f0c832e194a7191374f64fbd73281 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/268ec39d66eb44d387dd0a4ee6fef210 : SUCCESS in 7m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0ef9da944e3c47a7ba8989e097bd6a54 : FAILURE in 10m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/688e6204ae9e46959ac1b3575d9f7158 : SUCCESS in 1h 09m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6dffb2e890d74c33aaf38d55f7ce0735 : SUCCESS in 1h 23m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b3ab42108d7043748bcede761c9c3eaa : SUCCESS in 1h 06m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65fdcf6203034b35a58850f19ddc07d6 : SUCCESS in 41m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/032b8e14a2264ba18520df8783214724 : SUCCESS in 1h 10m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/43c6fee9b7f043f8aec58cd1b3203292 : SUCCESS in 39m 21s","accounts_in_message":[],"_revision_number":2},{"id":"937d9c89dbaccc2f6be300a600104797cb929e7e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-15 13:08:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d6fa24b91a14acb5b237b126e860459a2bb01f1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 13:14:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2caf45aecfc04bc2bfa6a90bf2cdbf52\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03a19619222e407dbd7d631e74a1602f : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58a438b3963f46aab6175dbbd046ab83 : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"853d568508b25606b12c49551c82a7770f2db856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 15:27:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06f8bfec74034ed0b41e180d341160ef\n\n- grenade https://zuul.opendev.org/t/openstack/build/d269fcfbfcc3414186ad63bc1c44f0a0 : POST_FAILURE in 58m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a78b7cf8e35a44dd9880bf68ec16fc1b : SUCCESS in 58m 24s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5bc0a5dabae747c5a53c2b3c8412a797 : SUCCESS in 1h 25m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0834849d78244fca64681cedf79ae5d : SUCCESS in 1h 32m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3568d0f09aa04964821e201018354d19 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7d542b084b346d9a968153af4db9a54 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7d577d43ee84b629b8333f5e76b056f : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a87fa57b22646ea883a154cd5c67f85 : SUCCESS in 6m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/27cc884a1f12423480fe0ba4b0249458 : SUCCESS in 14m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e49668b4672b49b18656293806aa85c2 : SUCCESS in 8m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/90cfb310d90f451a8c94e98ccecdd485 : FAILURE in 17m 12s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/955372dbc40b4272bb9de03b3ae2f7c2 : SUCCESS in 1h 13m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/09de5bb175174502aeb6f3de8fefc1c2 : SUCCESS in 1h 29m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/11d02d30448640f7aed74da69bbf2b88 : TIMED_OUT in 2h 06m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1f27cad1714f40fe957df55fc02c6d05 : TIMED_OUT in 2h 09m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17802b718c804cf68737d75eb7e0ef61 : SUCCESS in 52m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c66b9e00556946e68944e109785a9b04 : TIMED_OUT in 2h 06m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5fe35f4da4d74c18b767270843b06e73 : SUCCESS in 37m 53s","accounts_in_message":[],"_revision_number":2},{"id":"556cd0c948d73d5be693d6843794eb43941c1906","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-16 07:31:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f650920a9d0267b11ec282bce0a16f4bedbb7ea8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 07:45:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66d281b89ae94eee99637bd801a90167\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/85cefabef25e4058a558bcd506b374fc : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/203ab303da6a46c5bd36bb3904c89640 : SUCCESS in 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d222d4adcf890e4cf6fd47604763d45955c8123f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 09:16:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad4db159ded469c9f7bc0baabfe2135\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f144736eb1c457d844bc18bc9239c80 : SUCCESS in 1h 14m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/536cd30527474d6299c6ca17c5b37f4f : SUCCESS in 58m 26s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e3104abb2e7044908a3e66b752ee2a66 : SUCCESS in 1h 20m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9247bdbcc4824c2e88b59b949691b1d0 : SUCCESS in 1h 02m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/248c52400bcf4bfe80689b7e32fd5bcb : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a1ab06d066c4ba390e431169564a7bf : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2b904f047dd497e8d36c439309ed138 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a1211e786d2431dbd0b7b1dec9101da : SUCCESS in 5m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9e0930b7648b458daeb68293932551d0 : SUCCESS in 12m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/82d6ac8c65484221b1fab39185b97f9e : SUCCESS in 7m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5377cbbb65dc4f54b97b426c4d42d3f8 : SUCCESS in 7m 32s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9cd35b13cd62440ab5ed986a0a1f3d55 : SUCCESS in 1h 20m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e16ba9a3decd46afaf8cd682ee8ddcf1 : SUCCESS in 1h 24m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1396dceabc18442b8810a93e3095e136 : SUCCESS in 1h 20m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7ded763fc72e4146895a791fc0304e3b : SUCCESS in 1h 39m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7cbd255820545b3b59df0aecb253fd6 : SUCCESS in 51m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d81673d80ae74597b565103c5b9f90e6 : SUCCESS in 1h 15m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4b8014c4e5ca4dd99b10c37f648c5203 : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":2},{"id":"abc7758c062f7f6c8ff2f66055c561b06df95c51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 09:17:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"91721fcc5b909cc71cb0448b291d3f369deb2b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 11:53:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca2320952c3d4308ad0fd25739e3cfff\n\n- grenade https://zuul.opendev.org/t/openstack/build/ccb351b07a114223bb66fd4bb74e0211 : SUCCESS in 1h 06m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b16ecc5d779479db30a11454a04da0e : SUCCESS in 1h 41m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5af8c034116945f19e5df383a97f114f : SUCCESS in 59m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e67dfe73b3e5474383b78e8fe74d2ba2 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51ed710a494549aabb225cc2f6364081 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72845b23a6804dd98dc36e3d6e305bbb : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f636c6fe8e343b9a74c99902de9a524 : SUCCESS in 6m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e276cefd56394c7d9eeff6fe1bf7fa45 : SUCCESS in 8m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d6e52f1e85e94ea7b44688fe939756e9 : SUCCESS in 8m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8177ab741f8a4f97940a21ec1dcb69ef : SUCCESS in 1h 24m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2731ac86bd59484c97076573039af9ae : SUCCESS in 1h 17m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b30582f9103848afa1cdca95d34bb8b7 : SUCCESS in 58m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cc5a678312c4579ae22998953e857b5 : SUCCESS in 58m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9e3dd47e6b1e43aabae419c08c13e699 : SUCCESS in 1h 23m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9d5379c53a834a7192938efe34db8b4c : SUCCESS in 46m 43s","accounts_in_message":[],"_revision_number":2},{"id":"0ee90e3f17c23da0d5c5dc965657ec92943c79e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 11:53:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cd7c90a6a8a270849696bca3bffe56c85a765a66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 11:54:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a162060c54c64757814da0cb04367560\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f2e975aec448fe824b0a84817b6711 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"399438cd044f52915e202650db407eccd3e26c95","revisions":{"1019cc1f5c7258b9126ef50558f7c41d627db2dd":{"kind":"REWORK","_number":1,"created":"2022-07-13 18:30:03.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/25/849725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/25/849725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/849725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/849725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/849725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/25/849725/1"}}},"commit":{"parents":[{"commit":"7a42b7be1a9d126d69f8677cded8c3b7866fb246","subject":"Merge \"Refresh Glance example configs for Zed milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a42b7be1a9d126d69f8677cded8c3b7866fb246"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-13 18:28:03.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-13 18:28:03.000000000","tz":120},"subject":"Fix E741 issues","message":"Fix E741 issues\n\nIn 2.6.0, pycodestyle is going to find more E741 issues[1], which is\ngoing to break the gate by making \"tox -epep8\" fail.\n\n[1] https://www.flake8rules.com/rules/E741.html\n\nChange-Id: I94ee03a00dd01726de15a29424183faabdeba844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1019cc1f5c7258b9126ef50558f7c41d627db2dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1019cc1f5c7258b9126ef50558f7c41d627db2dd"}]},"branch":"refs/heads/master"},"399438cd044f52915e202650db407eccd3e26c95":{"kind":"REWORK","_number":2,"created":"2023-01-16 11:16:14.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/25/849725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/25/849725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/849725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/849725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/849725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/25/849725/2"}}},"commit":{"parents":[{"commit":"e9b40e1316b7e262887ddcb3d80067d5f7593540","subject":"Merge \"Release notes for Antelope Milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e9b40e1316b7e262887ddcb3d80067d5f7593540"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-13 18:28:03.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-01-16 11:01:06.000000000","tz":60},"subject":"Fix E741 issues","message":"Fix E741 issues\n\nIn 2.6.0, pycodestyle is going to find more E741 issues[1], which is\ngoing to break the gate by making \"tox -epep8\" fail.\n\n[1] https://www.flake8rules.com/rules/E741.html\n\nChange-Id: I94ee03a00dd01726de15a29424183faabdeba844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/399438cd044f52915e202650db407eccd3e26c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/399438cd044f52915e202650db407eccd3e26c95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
