)]}'
{"id":"openstack%2Fglance~850028","triplet_id":"openstack%2Fglance~master~I6cd8e0ad3e5055a224c9ccee388175766d52b44b","project":"openstack/glance","branch":"master","topic":"dnm-testing-git-review","hashtags":[],"change_id":"I6cd8e0ad3e5055a224c9ccee388175766d52b44b","subject":"DNM testing git thin/non thin pushes","status":"ABANDONED","created":"2022-07-15 16:26:10.000000000","updated":"2022-07-19 17:57:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f2638faaad10865380088419085a155d6eb09f4e","_number":850028,"virtual_id_number":850028,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-15 16:41:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 16:41:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d95a8616668f592c7fe07a1b3cc5a7844ab3d66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-07-15 16:26:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee386a4a05e3271f97cad3e1138f0503697dbc8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 16:41:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccb188ae930d4f06bc03bc863b265b9f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/df66ae1325484149b3fb507345623029 : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c12cdb1fb3d44cdb9cda623d3be686e : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3162cfa06630471a8155aca96038ebb5 : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f2638faaad10865380088419085a155d6eb09f4e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-07-19 17:57:39.000000000","message":"Abandoned\n\nThis pushed successfully which is what I wanted to see (or not see)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c69ec8475892efb6019cb3c37feb5f3a40c42e7","revisions":{"5c69ec8475892efb6019cb3c37feb5f3a40c42e7":{"kind":"REWORK","_number":1,"created":"2022-07-15 16:26:10.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/28/850028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/850028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/850028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/850028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/850028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/850028/1"}}},"commit":{"parents":[{"commit":"d6184d29dacd4cd6fcab6ac028826a69c445b772","subject":"Merge \"Release notes for Zed Milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6184d29dacd4cd6fcab6ac028826a69c445b772"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-07-15 16:25:41.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-07-15 16:26:02.000000000","tz":-420},"subject":"DNM testing git thin/non thin pushes","message":"DNM testing git thin/non thin pushes\n\nChange-Id: I6cd8e0ad3e5055a224c9ccee388175766d52b44b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c69ec8475892efb6019cb3c37feb5f3a40c42e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c69ec8475892efb6019cb3c37feb5f3a40c42e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
