)]}'
{"id":"openstack%2Fglance~850036","triplet_id":"openstack%2Fglance~stable%2Fwallaby~I04efbb7fd47b8b1e022d96c40011b6f03508db2e","project":"openstack/glance","branch":"stable/wallaby","topic":"bug/1979699-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-09 14:57:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I04efbb7fd47b8b1e022d96c40011b6f03508db2e","subject":"glance-cache-prefetcher: Set up access to multistore backends","status":"ABANDONED","created":"2022-07-18 15:42:45.000000000","updated":"2023-02-09 14:57:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":848992,"cherry_pick_of_patch_set":1,"meta_rev_id":"72983fa0e2c5f59eff42c758586d5890794ed32e","_number":850036,"virtual_id_number":850036,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2023-02-06 20:59:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2023-02-06 19:12:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-07-18 15:55:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-02-06 19:12:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-18 15:55:42.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-07-18 17:12:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-06 19:12:34.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"ae3c4c0f6c06874d0eaa187ea5fcc443979358e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-18 15:42:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"52c8a48aaaa42832ebe71ab86090c7cd04598803","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-18 15:55:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a0ddba6a032ff5a30f8e78b01c0568a059c1cced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 17:12:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb0d4560124348eda33e6d5434e90887\n\n- grenade https://zuul.opendev.org/t/openstack/build/c16d793917884f978e845ea43fd9e73a : SUCCESS in 1h 10m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13d49315fac34e17b2fc383ad16509a8 : SUCCESS in 1h 22m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fbcf753ec114bb29e3fe3f2fc417bfb : SUCCESS in 55m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30eb7d1dc4f04dd1b8ac923e2bc48bde : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75b6208464049bc95490bb146e3b10a : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c3068d3b1d1429f94662f7a2474608b : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e2bccf777d94adb8d1bc13f90782642 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/968613815c614b7cb76b4c63bbe1c92c : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b23c203aef74952866aae003e59b451 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f09f8f2a59594a099ee0dfd742aec187 : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d414c5ba21aa4f16ae380daf188f430d : TIMED_OUT in 30m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb50251059a647d38f5fe4034f5e5bdf : SUCCESS in 8m 57s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1dbdef4f485240ef9fcad9f16429a4e3 : SUCCESS in 8m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9e28f696ed3d4f99b10df43308f2f951 : SUCCESS in 1h 08m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/842c993b82e84ec7bdf2484bb44472b6 : SUCCESS in 1h 14m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/873528a4982b41699f5551f777a4e276 : SUCCESS in 1h 13m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2c4a06cf60654c31bff2d41b40fa366b : SUCCESS in 1h 24m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c984a5ea38a44bf9b3f2810b1049678 : SUCCESS in 59m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/700911ed13514bd5b106fb8afc328bb5 : SUCCESS in 1h 23m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/87ca7a99c4de431f945b3f5a9ae6250e : SUCCESS in 37m 12s","accounts_in_message":[],"_revision_number":1},{"id":"28884efd25d722369a0ecf589cddf189be821317","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-18 19:08:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b6add10ee69eddb0a8bbf406209a22df8c13c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 20:35:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2f1afd49dd34b2e9918fab17a0c3533\n\n- grenade https://zuul.opendev.org/t/openstack/build/26410013cde1448ab7d147c5683429a3 : SUCCESS in 1h 02m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e9ac258d4c5c4cef9997ffb22b4286c8 : SUCCESS in 1h 18m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38af04ac5e3741e182b77fcf2c3565d8 : SUCCESS in 54m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c85d47129a6442ba65d3e349f4fca21 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/510cb0b8387145cdb2bbd0dbca7404f1 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4ca44b2a9d3407f962d92445865503f : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/041e199d5fc84afc8153450755a886fa : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/219357c3ed754999a6bd0951a07371d5 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eccccd67c5e4971b8bc7336a1294aa6 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de725f6e7e1c4022a16b1f3c082ac900 : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5048a6e6ef37420ca6f482552398f638 : TIMED_OUT in 30m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a2ba93ef6064a8aa6b98b6c66afae2a : SUCCESS in 8m 01s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f856cd0ef05b48f28de08dcf26c3391b : SUCCESS in 5m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cb7288b6c04943f9906122930cecbcf2 : SUCCESS in 1h 25m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e1ff3e76729e4da8a961f79d9f4147b5 : SUCCESS in 1h 09m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0d1062db8cca4519b4604f45d6be7fd7 : SUCCESS in 1h 19m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8bb36f039ee64525b439a4dfe2db95e7 : SUCCESS in 1h 17m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f66bad9f9fd249ce9eb51921311fdccc : SUCCESS in 53m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c432aae80d514301b3a817276dd944db : SUCCESS in 1h 22m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c04c57bc33fa45d6851166d899b51140 : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4b35b545bae24cefdff1795796f1e17534f4dc9b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-19 18:53:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"da48a79401d03503abdcd42b356c65869849862e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 20:23:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c83bf50aff704fcfb2c22367c93ab45a\n\n- grenade https://zuul.opendev.org/t/openstack/build/a8dd7415969f43088ff7e7a2f6d2ee9c : SUCCESS in 1h 01m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a53dbd391d44fbd985e008a1ef22c6f : SUCCESS in 1h 19m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a36a2130c9fd40178c481358bdd5f233 : SUCCESS in 50m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/560128607a17468597f5dce96c71d833 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6984067bee964248ab7b1a2192ea80b9 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e26b284c770436abccde1c5357a0e5b : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbb3ea14f9d44ff5b17276568dea95ce : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b561ef155c945d1ab7f265e96e8e795 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c28bf13dd24bd1941e56117b81454c : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42b443d5197a420399e47d17aa1883da : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/468d94ba52bd40df9a9dbe1db02fdf93 : SUCCESS in 8m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37d553358e594dcf83171b750f3491a6 : SUCCESS in 7m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/94d27ae7a4034a7084d3dadf6c919df6 : SUCCESS in 5m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/326b8da2cfb34a4184ca28d02e34b2a1 : SUCCESS in 1h 11m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5191ce00dcbc4cd499239b0205ad7eb8 : SUCCESS in 1h 20m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b3610bd968ee4b1c9559a7513402c48f : SUCCESS in 1h 23m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c7a90fcdca8940a791fb9bed1bc35fed : SUCCESS in 1h 28m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8bda018fe2e54d51b4389e1be8404d2c : SUCCESS in 59m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1d464f7e2e6541a3bf0c9a35e5ca0bbb : SUCCESS in 1h 16m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/220312c2315d4452ae37c796010d062c : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":1},{"id":"65818bdf729f37cc2c33ced1d4a6afebc7477967","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-06 19:12:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4341220d820b1968595646a9055b63c724ae3ba6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 19:13:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41040ce653777b9e083f75ad8ff9aad32024ddc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 20:59:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82d8c9dd6d7c4ea7a700b9138973d6a8\n\n- grenade https://zuul.opendev.org/t/openstack/build/aa766f74869944b5ab8220f04159f7ac : SUCCESS in 1h 08m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a6f1a8f86a54580ac53602e12fbb2ea : SUCCESS in 1h 25m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/166d1e98fa4f4b5fb1d1537039ce855c : SUCCESS in 1h 06m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/279be744b9554325b03e44d3438e46f6 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f8bf05f74444bf810083b9c3f6aef8 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67c76ab4cbab42a1965fa65d7ea15cd8 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f6955397b864c8a8f942fed2b68ea03 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3855ce01b4b645d19785aadefd215226 : FAILURE in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bcd9914350f430d8cab667cabf7cd6b : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b560b0b7408417096677e95536a45e4 : SUCCESS in 8m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9df7d35c63044488acac4a3c02a2c7da : SUCCESS in 10m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4c700a1ca2f14e53a95443080260d41a : SUCCESS in 1h 40m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7f455144682d401991700e8d8f49f6aa : SUCCESS in 1h 28m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/869a0c94b868466d8d07a61b22fab8e7 : SUCCESS in 1h 02m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7040362421254255b5771ef5a60f3233 : SUCCESS in 1h 12m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/10d1709267e846a6b472bb1bcafdfb09 : SUCCESS in 26m 51s","accounts_in_message":[],"_revision_number":1},{"id":"72983fa0e2c5f59eff42c758586d5890794ed32e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-09 14:57:50.000000000","message":"Abandoned\n\nThis branch is too old, we will not merge this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9203ccc6fcad9388b3b36d8aff1fe08eef647433","revisions":{"9203ccc6fcad9388b3b36d8aff1fe08eef647433":{"kind":"REWORK","_number":1,"created":"2022-07-18 15:42:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/850036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/850036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/850036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/850036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/850036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/850036/1"}}},"commit":{"parents":[{"commit":"677c89c23631e9083261a1a18ed438d8966e0de2","subject":"Fix set_property_atomic() boolean type casting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/677c89c23631e9083261a1a18ed438d8966e0de2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-24 01:02:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 15:42:45.000000000","tz":0},"subject":"glance-cache-prefetcher: Set up access to multistore backends","message":"glance-cache-prefetcher: Set up access to multistore backends\n\n... because the command requires access to backend stores to pull image\ndata.\n\nCloses-Bug: #1979699\nChange-Id: I04efbb7fd47b8b1e022d96c40011b6f03508db2e\n(cherry picked from commit 6964e627d5b98c2034aea4bebb72cdab7549b6b5)\n(cherry picked from commit b4091752b4a771a54b5298d15729352ae9552e81)\n(cherry picked from commit 2f125d83293ab4bf8037770fe8f620bc3664d0c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9203ccc6fcad9388b3b36d8aff1fe08eef647433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9203ccc6fcad9388b3b36d8aff1fe08eef647433"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
