)]}'
{"id":"openstack%2Fglance~850561","triplet_id":"openstack%2Fglance~stable%2Fyoga~I795c751edd1403d23aa1de5cda194aada80d05b2","project":"openstack/glance","branch":"stable/yoga","topic":"fips-compatibility-stable/yoga","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-08-09 11:33:34.000000000","reason":"\u003cGERRIT_ACCOUNT_34120\u003e replied on the change","reason_account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"}}},"hashtags":[],"change_id":"I795c751edd1403d23aa1de5cda194aada80d05b2","subject":"Fix fips job","status":"ABANDONED","created":"2022-07-22 11:20:08.000000000","updated":"2022-08-16 12:26:34.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":847089,"cherry_pick_of_patch_set":2,"meta_rev_id":"4f7642a852bef20d60e92eb481c5bda064d77456","_number":850561,"virtual_id_number":850561,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-22 12:21:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2022-07-22 12:14:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-22 11:20:08.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"CC"},{"updated":"2022-07-22 12:14:52.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-07-22 12:21:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f50bb7e4138b34f8096be43528673071b86b821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-07-22 11:20:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4dccb0b351c4ef41833de76a90fce9a2246ddea4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-22 12:14:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"74ed36034cc9bfc41d4a8c88a11e8ac405fca265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 12:21:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ec22c59476b4a7b9c1c7014e01579dc\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7f02d210ccd0402baba4c664106cef37 : SUCCESS in 57m 48s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3df4e9a47334d5a8b14e63a272a9a62 : SUCCESS in 46m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c8da8a10654ddabc9841ca1f84acef : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e53df90ebe242cca4b78b0fd9d83d93 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/054bd029100448b79c4da1dccfd7bee0 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dcfb685cf524f5fa5072adc30ad9646 : SUCCESS in 6m 14s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/dc6b761e3d324951a9d85cc5180bf928 : SUCCESS in 11m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c7aee9384c314038b0d4fa218992585d : SUCCESS in 8m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d783bf128baa4c8ca6ddf41d436e1429 : SUCCESS in 9m 01s (non-voting)\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/3c1acefb785f4e29aa6519cce76262a8 : FAILURE in 38m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fb5c120e14f04051b401b67ebb2e1068 : SUCCESS in 28m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e0e51449314d98a63adeaaccf18bf3aaae4d900f","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-08-09 11:33:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17732ace4aa26eb8969ffa80a42a4773b010cbed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 12:54:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d69f48ec57db465c9cfa60d6150a584f\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b8183ea959fa40e3988c4abd5462f28f : SUCCESS in 1h 15m 12s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f592ad6178da47df847fd975435cf033 : SUCCESS in 1h 02m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/999147b1914d4b0aac854d6c4aa4b264 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d7d47a8de0041f99030c5a251852d9c : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e754b2edf8842a3a1ae3a14ce709cfb : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64efc7b771c547928ee477523f748583 : SUCCESS in 7m 25s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/053f0102c5a54872a7f6e20067d734fc : SUCCESS in 12m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/615403d13a9446d8a6b1e45af90bf8a9 : SUCCESS in 10m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/de15d5e5fc2246f7a84f8a69ac919ce4 : SUCCESS in 8m 48s (non-voting)\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/057ca9bfc8b04e408bc48d1aab6a3b1a : FAILURE in 38m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/da8af1a2df8940adba393a5205d3076b : SUCCESS in 47m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4f7642a852bef20d60e92eb481c5bda064d77456","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-08-16 12:26:34.000000000","message":"Abandoned\n\nAbandoned in favor of https://review.opendev.org/c/openstack/glance/+/853283","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3dfef22838b5169c4f49a32ab2a08b25e556bae","revisions":{"d3dfef22838b5169c4f49a32ab2a08b25e556bae":{"kind":"REWORK","_number":1,"created":"2022-07-22 11:20:08.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/61/850561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/850561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/850561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/850561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/850561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/850561/1"}}},"commit":{"parents":[{"commit":"b4091752b4a771a54b5298d15729352ae9552e81","subject":"glance-cache-prefetcher: Set up access to multistore backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b4091752b4a771a54b5298d15729352ae9552e81"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-06-21 19:36:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-07-22 11:20:08.000000000","tz":0},"subject":"Fix fips job","message":"Fix fips job\n\nThe ecdsa usage for ssh is supposed to be the default right now.\nAlso, a new role paramter nslookup_target is required for the fips role\n\nChange-Id: I795c751edd1403d23aa1de5cda194aada80d05b2\n(cherry picked from commit 05543914a774bb4a170b789f64e6945bcfbc4259)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3dfef22838b5169c4f49a32ab2a08b25e556bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3dfef22838b5169c4f49a32ab2a08b25e556bae"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
