)]}'
{"id":"openstack%2Fglance~850589","triplet_id":"openstack%2Fglance~master~I3a88ab5d2e67a1553f03b0b8ba44efbef976ffb0","project":"openstack/glance","branch":"master","topic":"py311","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-08-01 07:39:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a88ab5d2e67a1553f03b0b8ba44efbef976ffb0","subject":"glance-manage: fix crash under Python3.11","status":"MERGED","created":"2022-07-20 22:27:16.000000000","updated":"2022-08-01 07:40:55.000000000","submitted":"2022-08-01 07:39:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850589-py311","meta_rev_id":"87441dedc705215a3b0d8d6c99db4113d1b7e07c","_number":850589,"virtual_id_number":850589,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-01 07:39:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2022-07-26 10:50:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2022-08-01 06:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2022-08-01 06:02:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-21 00:18:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-26 10:50:46.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2022-08-01 06:02:43.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"f061ea6621fecc0f1d7e04972405053115c5c354","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-20 22:27:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae3aadab70e0c15a03aa34ee4f98d681b3aaa32f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 00:18:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9a259edf2be4f2397770ed121f32234\n\n- grenade https://zuul.opendev.org/t/openstack/build/0e44883cc21a4f31937032ed16036965 : SUCCESS in 54m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c92ba4c1ea264661b1f869ad71f500f6 : SUCCESS in 1h 16m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b14a1a90c6b42f0bda9147c42954763 : SUCCESS in 49m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ec1f69577c4dcaa7dda8ed84be89e0 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a70bb46e6ac48f68591cfcaf9a29737 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f25e9608c2804f3693608b1f3dfed037 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8271b2416381410c8ba65127abc66ba3 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48023f917f5048b5ad81f4914006a367 : SUCCESS in 5m 36s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8b891dd05374487795ae591ebddf2cff : SUCCESS in 13m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f47aeab0a4374e42972f915ca9a4fd90 : SUCCESS in 8m 59s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7e2b210ed8564abba44cfd956c2037fb : SUCCESS in 10m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e97a8b1df7b84bc28465c42a0ac92685 : SUCCESS in 5m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4f2febb9a02c4a66b9b134b7074b6c4e : SUCCESS in 1h 38m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f3232eea823644cba6a2556023c98774 : SUCCESS in 1h 33m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4a6a619402264b97b8013f7b3a3c5989 : SUCCESS in 1h 35m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/aab699cbf267496ea681de1a255739a0 : SUCCESS in 1h 49m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f33486a970242ff8ea2814db213ecef : SUCCESS in 1h 02m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cd8cdcf605d442c1a1015eb4fa20d23e : FAILURE in 1h 38m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/12622713df6e433c992c537dcd2127ad : SUCCESS in 32m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8e69f97621d09de5c633c07f5244854ef740d5af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-21 00:33:56.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"01fa228ff36ff2c5a917d87114026b61a2eed162","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-21 00:34:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0f29760c608871a5723be090dd7ceea5ca5c7a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 02:12:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25a581a2120e467e83831d785c46e5fd\n\n- grenade https://zuul.opendev.org/t/openstack/build/833f60e5be8542549887c6db4a0367e0 : SUCCESS in 1h 05m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f0d9715bde7438eaa599a013b6efc9c : SUCCESS in 1h 23m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbb6982cd7704f44a7f5da42e74d0649 : SUCCESS in 42m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d77114f5bb5042dba483e8a213e356f2 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20f700d3c5f6438bb4fd6491adbe7a0f : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38d73b303640450aa2cc2f2dcd584c97 : SUCCESS in 9m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c919c539b964137b5dc0a65139788f3 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4114661b01d42f7a8b1e058b72f50b6 : SUCCESS in 6m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/50ab46a6579f4ea983033bbaa7d00272 : SUCCESS in 12m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/63627fb2b8d34dbd8b0c3f2b43595ccd : SUCCESS in 9m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/577225486f534dfbb22c4556f5612911 : SUCCESS in 9m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3fba348269054d3daaa30598222f80a5 : SUCCESS in 5m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b3cd779d4e6d48a294d3aed96f2f9199 : SUCCESS in 1h 15m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d9d1c2af5a60421aa4c86eed70c7e0cc : SUCCESS in 1h 00m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b061729efb794a09b189c9de61aff5ed : SUCCESS in 1h 11m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/33b9f149dbd24f01a74f73217abc5caa : SUCCESS in 1h 27m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eaa0165d380847b6a8b183a27a1bc948 : SUCCESS in 41m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3c1a40712d7a42b59b4ef8992406a866 : SUCCESS in 1h 14m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/296b0ddf2b6f4165b7f22f91b1a42b97 : SUCCESS in 32m 12s","accounts_in_message":[],"_revision_number":2},{"id":"95c3fc008069f6f31dafa6e30b58239232346cba","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-07-26 10:50:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0bb9958369f57e3cae9db91efefd657e5321e3b4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-26 15:23:46.000000000","message":"Topic bug/1982426 removed","accounts_in_message":[],"_revision_number":2},{"id":"34ddbd332ef1a946101cfc4ae8c1d9c65e281b8a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-26 15:23:50.000000000","message":"Topic set to py311","accounts_in_message":[],"_revision_number":2},{"id":"b4c66b49aa17b91ede1fff6df0ccb73d9128b7a6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-01 06:02:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"144c959766964364f10cf31179c397689ee70447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 06:03:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7491247a3da27dbc527756da3c6307f0bda436a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 07:39:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acda8af5c6d44528b9fea7e452f14339\n\n- grenade https://zuul.opendev.org/t/openstack/build/7834669bbb8d448b9c5b444d001c48d1 : SUCCESS in 52m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3c5d054cf7942c0a94acd8a3cfa362d : SUCCESS in 1h 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08e3645ce4aa45bbb12c7876d499525b : SUCCESS in 47m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69755fbd54da4b38b1867b7cdf1e315d : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ea4d3ced42f4061b9973189dc083358 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b68d223890c847b19e2481c6cf828437 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adc683d6f6ee40eba66e3a9674cb7e3d : SUCCESS in 6m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a665095f826a40db9016f6e73cf83503 : SUCCESS in 8m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6d2d9c701eae4fcb983c77af349f8541 : SUCCESS in 8m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6731cc75cc8f461c8f95182dddc68210 : SUCCESS in 1h 35m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f65dc37675234afe8e7d566a6099612c : SUCCESS in 1h 28m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef329f74df864409b1c2cccf2ad371df : SUCCESS in 46m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eb10412a76a74ac083ef6f2af77fb805 : SUCCESS in 1h 16m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2c55c2705fa84050b26f032e98ebf228 : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":2},{"id":"0d6c1bc86ca0448c2d74b0b8d8be4dbf30cf595c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 07:39:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"87441dedc705215a3b0d8d6c99db4113d1b7e07c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 07:40:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d656ca5aedf467fbe6629e45e9a858d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f9916b36e0433c89c2a3cf370eebec : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"76abb227addf1ae44813527084d1ceb7f36af606","revisions":{"481a1728f582c0e8499c4763c5774dcd458943b7":{"kind":"REWORK","_number":1,"created":"2022-07-20 22:27:16.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/89/850589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/850589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/850589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/850589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/850589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/850589/1"}}},"commit":{"parents":[{"commit":"00f453372ccd4faf99e9a0bf527762a131968ad0","subject":"Merge \"Immediate caching of an image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00f453372ccd4faf99e9a0bf527762a131968ad0"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-20 22:16:20.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-20 22:27:08.000000000","tz":120},"subject":"glance-manage: fix crash under Python3.11","message":"glance-manage: fix crash under Python3.11\n\nIn Python3.11, the argparse module became stricter. While the following\ncode was fine in previous versions, it will now raise an\nargparse.ArgumentError because we added the same subparser twice:\n\nimport argparse\nparser \u003d argparse.ArgumentParser()\nsubparsers \u003d parser.add_subparsers(title\u003d\u0027Commands\u0027)\nsubparsers.add_parser(\u0027foo\u0027)\nsubparsers.add_parser(\u0027foo\u0027)\nsubparsers.add_parser(\u0027bar\u0027)\nparser.parse_args()\n\nThis mistake was silently ignore in Python3.10, which explains why this\nwas never an issue for glance-manage.\n\nChange-Id: I3a88ab5d2e67a1553f03b0b8ba44efbef976ffb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/481a1728f582c0e8499c4763c5774dcd458943b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/481a1728f582c0e8499c4763c5774dcd458943b7"}]},"branch":"refs/heads/master"},"76abb227addf1ae44813527084d1ceb7f36af606":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-21 00:33:56.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/89/850589/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/850589/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/850589/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/850589/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/850589/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/850589/2"}}},"commit":{"parents":[{"commit":"00f453372ccd4faf99e9a0bf527762a131968ad0","subject":"Merge \"Immediate caching of an image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00f453372ccd4faf99e9a0bf527762a131968ad0"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-20 22:16:20.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-21 00:33:34.000000000","tz":120},"subject":"glance-manage: fix crash under Python3.11","message":"glance-manage: fix crash under Python3.11\n\nIn Python3.11, the argparse module became stricter. While the following\ncode was fine in previous versions, it will now raise an\nargparse.ArgumentError because we added the same subparser twice:\n\nimport argparse\nparser \u003d argparse.ArgumentParser()\nsubparsers \u003d parser.add_subparsers(title\u003d\u0027Commands\u0027)\nsubparsers.add_parser(\u0027foo\u0027)\nsubparsers.add_parser(\u0027foo\u0027)\nsubparsers.add_parser(\u0027bar\u0027)\nparser.parse_args()\n\nThis mistake was silently ignore in Python3.10, which explains why this\nwas never an issue for glance-manage.\n\nCloses-Bug: #1982426\nChange-Id: I3a88ab5d2e67a1553f03b0b8ba44efbef976ffb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76abb227addf1ae44813527084d1ceb7f36af606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76abb227addf1ae44813527084d1ceb7f36af606"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
