)]}'
{"id":"openstack%2Fglance~851335","triplet_id":"openstack%2Fglance~master~I5636dc01592475109e0975e7227902990ea15ee0","project":"openstack/glance","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I5636dc01592475109e0975e7227902990ea15ee0","subject":"Imported Translations from Zanata","status":"MERGED","created":"2022-07-28 04:28:53.000000000","updated":"2022-08-01 06:13:14.000000000","submitted":"2022-08-01 06:11:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851335-zanata/translations","meta_rev_id":"f95209c857a7ffc099d6e1535b535163852d8bc9","_number":851335,"virtual_id_number":851335,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-01 06:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-08-01 06:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-08-01 06:01:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 04:37:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-01 06:01:32.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"3f9b778c0b857b5db51a5d9608204b7e05bbda48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-28 04:28:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02caea94166a431cd702abcef323c12e7ec29d89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 04:37:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70aa0a9048224536b292068a06030679\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8f0a37073b24b9fb9a35414f5c5622e : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4853836f221648fbbbdb287711baf280 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2e709f4136b48a585ff4b9b24cff5ea : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"dd134d5576e0d23fdcadb596b2bf5d98e9bcc7db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-29 03:39:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"23f5111079dd3b120be339567b34c553c2145ee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 03:53:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39f1ba9e9dd84353af554244566937f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4634eb2b92447dd953ef5dd54fbc850 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/249bcbc35f1f400985fefc2b1bb7168a : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34569f658d5348ffbb523558250a56d5 : SUCCESS in 8m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be9ab60e41dd4148b03bcf85c87724a0 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e52411ce7948e88be519c192267a29 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b1a2393f3e74696bf1ea57e7214e64b : SUCCESS in 7m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9cb86d0ff6e8465ea29c4646b8086cd1 : SUCCESS in 12m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d60fd781955048a18d1bdaba120b4941 : SUCCESS in 10m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9c7e312668bf488fa0ca821d75dcb79a : SUCCESS in 10m 40s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/173c2fb079cd45109e0b1a06ddc11237 : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":2},{"id":"187fde2272fca64c976afab29319e1a2ad1e7c42","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-01 06:01:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"517b567634302c17c4b59cae029d87346e953d57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 06:01:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4f5ffa4126da52c8143a7c559b70598fc9e9a2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 06:11:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bda84550e8324ea1a00427f940ead11b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/755052fe9b9a4f24b6abfdd3acb2ec33 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/360953405982420cb8dd70ad5f4b541b : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2bca09d5a274546ad5da800d593877a : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4cc835af9c4809a474bad7df2bd9b8 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f59a431e1e642e8a35b8d7dc7bbcd2a : SUCCESS in 6m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/39c89cdd240d4e90902f05d66618674e : SUCCESS in 8m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a914e7a1c3b4dc1af8bd1e9cf38fd38 : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":2},{"id":"408dbcd74557c706a77ed14bcf253ecf4e3f37a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 06:11:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f95209c857a7ffc099d6e1535b535163852d8bc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 06:13:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fcb4831a1794917b3f70453961c0400\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b749ab5970fd4f4494a587ea47b32a76 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e40a267d827484daf669e96e357a4f0 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f70970b8e43d773374e1e85a7a869b5e1cb63669","revisions":{"4dfa1579a3bbd2735501c5210b28598ea6820328":{"kind":"REWORK","_number":1,"created":"2022-07-28 04:28:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/851335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/851335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/851335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/851335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/851335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/851335/1"}}},"commit":{"parents":[{"commit":"00f453372ccd4faf99e9a0bf527762a131968ad0","subject":"Merge \"Immediate caching of an image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00f453372ccd4faf99e9a0bf527762a131968ad0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-28 04:28:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-28 04:28:51.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I5636dc01592475109e0975e7227902990ea15ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4dfa1579a3bbd2735501c5210b28598ea6820328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4dfa1579a3bbd2735501c5210b28598ea6820328"}]},"branch":"refs/heads/master"},"f70970b8e43d773374e1e85a7a869b5e1cb63669":{"kind":"REWORK","_number":2,"created":"2022-07-29 03:39:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/851335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/851335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/851335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/851335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/851335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/851335/2"}}},"commit":{"parents":[{"commit":"1ef06ef08ba177bd9ea93c25f96fa13aa6a2e9de","subject":"Enable Image cache tempest testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ef06ef08ba177bd9ea93c25f96fa13aa6a2e9de"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-29 03:39:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-29 03:39:27.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I5636dc01592475109e0975e7227902990ea15ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f70970b8e43d773374e1e85a7a869b5e1cb63669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f70970b8e43d773374e1e85a7a869b5e1cb63669"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
