)]}'
{"id":"openstack%2Fglance~851947","triplet_id":"openstack%2Fglance~master~Ib6567f99b93c2a2fd024dd72167ecf0956011373","project":"openstack/glance","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ib6567f99b93c2a2fd024dd72167ecf0956011373","subject":"Imported Translations from Zanata","status":"MERGED","created":"2022-08-03 03:44:10.000000000","updated":"2022-08-08 06:42:21.000000000","submitted":"2022-08-08 06:41:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851947-zanata/translations","meta_rev_id":"d6d150309fdf9bb0d4fde5138e1ba5ccd92188d3","_number":851947,"virtual_id_number":851947,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-08 06:41:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-08-08 06:26:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-08-08 06:26:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-03 03:57:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 06:26:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"93a33eb919aa7c4b2cbd3630553466bfbf96ca2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-03 03:44:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d75c3d5fd56a9dcb8c0e12da5fa326168f4c15f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 03:57:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b1d3469f1a4404e93859565aab4d339\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd3ed2fe3ef2407cb221b012ee2c2a90 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f66636d3be94f8c92f52139645a4806 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ee3995f37d3444aa456e6f10b1e9a40 : SUCCESS in 8m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da8a2d4225c347899f4476eeebed5b1e : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d01d4bab9d744a44a4c5abd3bade9c04 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53c736501c544d2f84ed2bafd5e3e2a2 : SUCCESS in 7m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0ffe5dd4c5194ea5a375f1189f9adbb1 : SUCCESS in 11m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b26bbafb636c42e7b9ef4a2bf12a0457 : SUCCESS in 10m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/acb0c1ed0c064418b42589976592ee82 : SUCCESS in 10m 43s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/524e92e0def14c2fbf5d54a747bd4008 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"418f97448c1f7913e0360667ed2fe88635dff03f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-08 06:26:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e09a02f3ae5d3df63c7814b98a39ccf3122ec5cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 06:26:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14086ed57876f69a5efa314dbccaa8d2bd046c9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 06:41:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffe671e47c5549ccb8dcb4baa257a9db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6724209f49be4f93ba12170428a53f3f : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14c034e615e74e4bb9e0e7b162557902 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5b9cf1e1fa24f1b9d87c004343fdd05 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e435b520df4f20b641420a037d25df : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/868a5bf4d7304cca9a4aad136b073977 : SUCCESS in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8c92718a3ac0440d84b582245da5098d : SUCCESS in 8m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/40fc13e577ad44b4ba9591963248274d : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"29b51dee4082a0b2ff927295a26426bd0f5e06d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 06:41:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d6d150309fdf9bb0d4fde5138e1ba5ccd92188d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 06:42:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57a3447c385d442f849e181697677bc8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d47215db7f3b453f9eb9c481d294bc67 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df9a19de074f40e2a47663524f33feb4 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7204126777f60fa9f7817741325fef95192bf041","revisions":{"7204126777f60fa9f7817741325fef95192bf041":{"kind":"REWORK","_number":1,"created":"2022-08-03 03:44:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/851947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/851947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/851947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/851947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/851947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/851947/1"}}},"commit":{"parents":[{"commit":"d6308c24a195dff83067341e2965170a46145260","subject":"Merge \"Update exception for property protection file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6308c24a195dff83067341e2965170a46145260"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-03 03:44:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-03 03:44:05.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib6567f99b93c2a2fd024dd72167ecf0956011373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7204126777f60fa9f7817741325fef95192bf041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7204126777f60fa9f7817741325fef95192bf041"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
