)]}'
{"id":"openstack%2Fglance~852380","triplet_id":"openstack%2Fglance~master~Ib03fefbb6034a7e7d2ff1e16430d28962c2ca355","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-08-08 10:08:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib03fefbb6034a7e7d2ff1e16430d28962c2ca355","subject":"Mock getaddrinfo in cinder store tests","status":"MERGED","created":"2022-08-08 07:58:59.000000000","updated":"2022-08-08 10:09:48.000000000","submitted":"2022-08-08 10:08:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852380","meta_rev_id":"a37c26fc2e8cb960a59afc85746f6960d27338d1","_number":852380,"virtual_id_number":852380,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-08 10:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-08-08 08:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-08-08 09:07:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-08 08:37:25.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-08-08 08:55:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60abde6fdbebaca1b1c08386203bf2d93bd327d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-08 07:58:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e88adeb1c2c456aaedda651331c26aec53f30a7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-08 08:37:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45c402e0686261249a42a65c4d93e22f17cf3006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 08:55:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf7be69572384bdf9511e1aba8a57a11\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49d1ba5b3b73434cb7e6a2bd56ea7333 : SUCCESS in 54m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d92a9fabb3784835a2a0b1167f5632c3 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35ad0adfca1746ca8258d7e72fff9adc : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfc987249dca43b683f94de2e7ec3f49 : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75e47d53a13e488e9b5a449ae8e262ef : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d13339b37f52456686335e11133a87fa : SUCCESS in 7m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/617044dae18a4ae2a8d55d455486cefc : SUCCESS in 12m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f859a0957a74e3a96603c64acd253c6 : SUCCESS in 10m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ef598b130d564d5784c243b34a958b4e : SUCCESS in 10m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1fc7ad647ecb4db5a23bea421702c864 : SUCCESS in 7m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/78caa826ccb14907a3ed5d1a08db78fc : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6dd1c37e4d5e6ad9e3a2303b303a22434f842b5c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-08 09:07:41.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c7e19158d79b1dfdc1e0c202e9242fe013b5a2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 09:10:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52cabd2eb76d6da0d565d6c2624fa231e62d4cbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:08:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ded94e3b25d46039a41168b2b774314\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/462ac043ed89487688e57c27df5ea9e2 : SUCCESS in 57m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05039f43b38c412cbee2f13ba260fcf1 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1aade3e7b7054aad878b41b2d6d87173 : SUCCESS in 10m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4be566549d2c4940b66cddbae99933a4 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5894e1866af346ea9db72dd144b4d2c5 : SUCCESS in 10m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ffed7d0d59c4afca58752748f7960cd : SUCCESS in 13m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b52d2a8c9f9b447090e85d930ebef6a3 : SUCCESS in 13m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c3200ad3b0af4cf4b4e198cd89e4cca0 : SUCCESS in 39m 03s","accounts_in_message":[],"_revision_number":1},{"id":"79f29b593441ae3ac339dbca2a2adb4e83cd4aae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:08:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a37c26fc2e8cb960a59afc85746f6960d27338d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:09:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eedd96a169947cc93822707e65a8741\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a047f49e16848c2b50cba7050663c6f : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f40ac40c2c40d57637312959b680121a9c89949a","revisions":{"f40ac40c2c40d57637312959b680121a9c89949a":{"kind":"REWORK","_number":1,"created":"2022-08-08 07:58:59.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/80/852380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/852380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/852380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/852380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/852380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/852380/1"}}},"commit":{"parents":[{"commit":"7204126777f60fa9f7817741325fef95192bf041","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7204126777f60fa9f7817741325fef95192bf041"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-08-08 07:47:26.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-08-08 07:59:14.000000000","tz":0},"subject":"Mock getaddrinfo in cinder store tests","message":"Mock getaddrinfo in cinder store tests\n\nWith change Ic79815972e654a8bfe2775f57c68cfa0bf115e2f, we started\nusing getaddrinfo method from socket lib to fetch the system\nIP address (v4 or v6) and pass it to os-brick.\nThe test works for most systems since a valid IP is returned but\nfor some deployments, it fails with the following error.\n\nraise socket.gaierror(socket.EAI_NONAME, \u0027No address found\u0027)\n\nIn any case, it is a good practice to mock it in tests.\nThis patch mocks getaddrinfo in the current cinder store tests.\n\nChange-Id: Ib03fefbb6034a7e7d2ff1e16430d28962c2ca355\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f40ac40c2c40d57637312959b680121a9c89949a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f40ac40c2c40d57637312959b680121a9c89949a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
