)]}'
{"id":"openstack%2Fglance~852504","triplet_id":"openstack%2Fglance~stable%2Fyoga~Ibd9563fb854f888b4a2f5a5dcf6abce0cfab0f04","project":"openstack/glance","branch":"stable/yoga","topic":"818966-stable/yoga","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-08-11 15:37:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd9563fb854f888b4a2f5a5dcf6abce0cfab0f04","subject":"Fix tests for logging connection info","status":"MERGED","created":"2022-08-09 14:21:14.000000000","updated":"2022-08-11 15:38:40.000000000","submitted":"2022-08-11 15:37:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852504-818966-stable/yoga","cherry_pick_of_change":818966,"cherry_pick_of_patch_set":4,"meta_rev_id":"7363dcbfc72208c7ec830466f97c7e4840b87f21","_number":852504,"virtual_id_number":852504,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 15:37:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-08-11 14:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-08-11 14:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-08-11 14:52:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-09 15:36:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-11 14:34:41.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-08-11 14:52:34.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"2235486e5d7aebe40a4acc4978267e7f19a009b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-09 14:21:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"befa60240de46621940ea128e68029a1ebf45639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 15:36:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fab14ea8ff6c45a0bc98af37e8e34f11\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/692483d5989a49b38102a907d2c65728 : SUCCESS in 1h 08m 12s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2d9e555de9b4cfcb797a5339d24b45e : SUCCESS in 55m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4968855bfd04d8bb9a74f5a5cb8d529 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c74ffb906244a3a8c8d2c2e4f020de1 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb542f3559be404ca7b54cd928586c87 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c174ce8583164020b1d31d5e770f53a9 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/a4b882851a5f4c2a971f3cff59603c6c : SUCCESS in 11m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dcbc744fcf1a49fda7d45c5359d66ecf : SUCCESS in 9m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c1dcbe53dd304dab8c23cfba8ffca5f8 : SUCCESS in 9m 00s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c04d46f671214b04b79a1c951ba94819 : SUCCESS in 4m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/929859a4f8484469a0050447895d8604 : SUCCESS in 30m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3f607a18e99d6199db777ac5c917688a9e2389c1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-11 14:34:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7b804e301c2f0e46bdd681add7775ac0b2a0a082","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-11 14:52:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3248e80b28f33179153e14878f3ab73e29048f2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 14:53:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64458e539921079ba464a61bee8db2b042c8134e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 15:37:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8c846bc26944e75b6d95b391ff06ee9\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c80c4187bcbc478d8a449ca56b0fb98b : SUCCESS in 43m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/387be76457224ffc94889c5f2e23fdda : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3736952cae04ea1950c8fcbca36e292 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bc0bffab40040bfba25d781c53febc9 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9cdd10efc654156a851a6f6d6487874 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec99386f548a4af18326cd6902c91f8d : SUCCESS in 8m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea835c07847348d69b03ee5d884b7c63 : SUCCESS in 9m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/134a062f488d4190b588929d464229f3 : SUCCESS in 28m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6d06b91778d05102a2f142c824f27a777a9d3c8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 15:37:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7363dcbfc72208c7ec830466f97c7e4840b87f21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 15:38:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eade1df0bee4cd9ac83678d660c0ac4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6d438b773714db08b7df3e45dc469ed : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20110a7d0c310eac81635d1f8c60bce423418820","revisions":{"20110a7d0c310eac81635d1f8c60bce423418820":{"kind":"REWORK","_number":1,"created":"2022-08-09 14:21:14.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/04/852504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/04/852504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/852504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/852504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/852504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/04/852504/1"}}},"commit":{"parents":[{"commit":"b4091752b4a771a54b5298d15729352ae9552e81","subject":"glance-cache-prefetcher: Set up access to multistore backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b4091752b4a771a54b5298d15729352ae9552e81"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-11-23 16:01:44.000000000","tz":-300},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-08-09 14:21:14.000000000","tz":0},"subject":"Fix tests for logging connection info","message":"Fix tests for logging connection info\n\nWhile logging connection info in the glance cinder store driver, we\nneed to mask passwords for which we use oslo_utils\u0027s mask_dict_password\nmethod.\nSince this is not mocked in the glance functional tests, we encounter a failure\nin cross-glance-tox-functional job in the change[1].\nThis patch adds the required mocking to pass the tests.\n\n[1] https://review.opendev.org/c/openstack/glance_store/+/818722\n\nChange-Id: Ibd9563fb854f888b4a2f5a5dcf6abce0cfab0f04\n(cherry picked from commit acc6f624cdb1b2e4a3900e7d226bbc5156e8fa17)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20110a7d0c310eac81635d1f8c60bce423418820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20110a7d0c310eac81635d1f8c60bce423418820"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
