)]}'
{"id":"openstack%2Fglance~852505","triplet_id":"openstack%2Fglance~stable%2Fyoga~Ib03fefbb6034a7e7d2ff1e16430d28962c2ca355","project":"openstack/glance","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-08-29 07:14:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib03fefbb6034a7e7d2ff1e16430d28962c2ca355","subject":"Mock getaddrinfo in cinder store tests","status":"MERGED","created":"2022-08-09 14:21:52.000000000","updated":"2022-08-29 07:15:10.000000000","submitted":"2022-08-29 07:14:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852505","cherry_pick_of_change":852380,"cherry_pick_of_patch_set":1,"meta_rev_id":"f59f7a71aa2894512de4063ebedde062d0c5dda8","_number":852505,"virtual_id_number":852505,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-29 07:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-08-29 05:56:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-08-11 14:35:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-08-29 05:56:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-09 15:35:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-11 14:35:30.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-08-11 14:52:50.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-08-29 05:56:43.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"b25f9e57230535b558e86e5d23846cafa58338c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-09 14:21:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"659a55508683cfdb59b7090c79248478f9e78df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 15:35:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2ec9bc55a79488e90dc501a2a8a2048\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c914ea0688ac45b989fc0d86c2020002 : SUCCESS in 1h 12m 12s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43dab67202fd4a9ea9f8ddea71375a9b : SUCCESS in 41m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28c05980da742c88cde259e132949f1 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7db502fa175e49fda0310f773a8ce6c8 : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf1ab029623e463f9505b1deec728d31 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c2ab2b3a5f48adab4ffe6850b25a04 : SUCCESS in 6m 13s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/99d5f3374ecf492a883625d7bf1009cd : SUCCESS in 15m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c9d3f8e03aa4e32bbd7860c0d0f40c4 : SUCCESS in 9m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ca74b033b1cc4e81904db121d2447f4e : SUCCESS in 9m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ab6e3b3dcdef402ea9d26df39aece8f6 : SUCCESS in 5m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/754cc225642e4e6bab48ce53297dc389 : SUCCESS in 32m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b3a290a44f25688b6a8a4a867025e6e60677e310","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-11 14:35:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7c794d458ff7c6715bcbe99d5fdb9912f8a1cf5b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-11 14:52:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a18a5ef95d4c62f193dff569e67919455f1beb11","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-29 05:56:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7c5b5559001ef9b43e231823e735d9223dc4822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 05:57:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cc2078dc92db19ecf13c3735cd45fafe6ca36a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 07:14:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327764e4a26a468fbdd53a4f36232e5c\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81e49f32021944a9af83739d30c56470 : SUCCESS in 1h 00m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08dbc60403ac4873be2d5a684f934c90 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2388349391e843f2b181349c3d13332b : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15a2e02b3a8b413490b5299ff167fe72 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f1e1a15476432585415846406a33ed : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c7496f4169ef407fa7790b3abae0541f : SUCCESS in 11m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0b63a7185fb74c4a819fde7f2ac230f5 : SUCCESS in 8m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/82d5fc15dad244408b4205582ae066db : SUCCESS in 43m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9c1606458177c3f1f6b6e8a3d849c2df8228d92f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 07:14:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f59f7a71aa2894512de4063ebedde062d0c5dda8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 07:15:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cffaffbf6a814e54a425d7ab95d31b71\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db9e38b2956e44adb4bd70bef5c83e25 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3ae67fa1925375fef6af44e73db201db4fe359c","revisions":{"f3ae67fa1925375fef6af44e73db201db4fe359c":{"kind":"REWORK","_number":1,"created":"2022-08-09 14:21:52.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/05/852505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/05/852505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/852505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/852505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/852505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/05/852505/1"}}},"commit":{"parents":[{"commit":"20110a7d0c310eac81635d1f8c60bce423418820","subject":"Fix tests for logging connection info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20110a7d0c310eac81635d1f8c60bce423418820"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-08-08 07:47:26.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-08-09 14:21:52.000000000","tz":0},"subject":"Mock getaddrinfo in cinder store tests","message":"Mock getaddrinfo in cinder store tests\n\nWith change Ic79815972e654a8bfe2775f57c68cfa0bf115e2f, we started\nusing getaddrinfo method from socket lib to fetch the system\nIP address (v4 or v6) and pass it to os-brick.\nThe test works for most systems since a valid IP is returned but\nfor some deployments, it fails with the following error.\n\nraise socket.gaierror(socket.EAI_NONAME, \u0027No address found\u0027)\n\nIn any case, it is a good practice to mock it in tests.\nThis patch mocks getaddrinfo in the current cinder store tests.\n\nChange-Id: Ib03fefbb6034a7e7d2ff1e16430d28962c2ca355\n(cherry picked from commit f40ac40c2c40d57637312959b680121a9c89949a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3ae67fa1925375fef6af44e73db201db4fe359c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3ae67fa1925375fef6af44e73db201db4fe359c"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
