)]}'
{"id":"openstack%2Fglance~852611","triplet_id":"openstack%2Fglance~master~Ifff797387ec06e8a1c0da13ebd722df0b993d672","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"34665":{"account":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"last_update":"2022-08-11 16:31:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifff797387ec06e8a1c0da13ebd722df0b993d672","subject":"remove expired notes for Multi Store feature","status":"MERGED","created":"2022-08-10 07:01:15.000000000","updated":"2022-08-11 16:32:30.000000000","submitted":"2022-08-11 16:31:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852611","meta_rev_id":"0cfeac6cc92b7b27eb0b7a0e99417455f564c393","_number":852611,"virtual_id_number":852611,"owner":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 16:31:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2022-08-11 10:04:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2022-08-11 15:03:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-10 07:08:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-11 10:04:26.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"fced78dc31c0f3d1acfa58f4c4abff93a5093157","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2022-08-10 07:01:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"366922250c8299f81b6bc94a370d3781eecde528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 07:08:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ccd99bf92f4e01bed2d7b1edd7e902\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/65239b5437fd4c7ab90bbc772b4794e0 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff97138ab7d34246a63b28456c579d98 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ae851e925a74785b84978da8ccdfb42 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"86aae0c057ab40650472d3abcdc2614f45419510","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2022-08-10 07:11:43.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"380d56cc940cd845c476f6cf42e5c48bc048c6f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2022-08-10 07:12:11.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f6f0c90b9e2fdcfd10ded0b9d6b513ed8a53d41e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2022-08-10 07:19:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6308e41c33b3272e7b3adef4aba70caa580049c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 07:31:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e44ac1999c53471a9708c81b07f0c81b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0e6d32f0d8b04ef7816db9bbcedf3fce : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b8e05c29c4497f9fdfb3bfbf1bb21e : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbdecbd908424d568589ea476395d1d7 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":4},{"id":"df86a2e4fc8a8a6e06231ebb1e3a91c30552d8c6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-11 10:04:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"29f9825a772f20c106cc9909f5006260b9edccee","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-11 15:03:56.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bf543243a7b56fd882b1e5553c32d1a9e4ae33c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 15:04:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0cba3cb18c5960636f830a94e6d4c8e126fc8410","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 16:31:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e001ae83e6e04d168bbbe97612c7f83c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/acc4d51f53a64fcbb4c236e27fafcbc7 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18f243701c0c462aaef2c5a3704aa6dd : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb985846cb7451ea5261e20ceb125ab : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":4},{"id":"4804f668913bb8009e9abf637aadbb04fb363d9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 16:31:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0cfeac6cc92b7b27eb0b7a0e99417455f564c393","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 16:32:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa9516c32ea34929bd483bcf4554f50c\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d750b16b3e204bf6859e19d4089e4856 : SUCCESS in 45s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73c63fa39ed64c6aabdeca49510853c6 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cb60c1d2facde2f318ec30c8b9cf63844eaccf92","revisions":{"c84197fc8ad21f546db6675a50198e5b3a1b51bb":{"kind":"REWORK","_number":1,"created":"2022-08-10 07:01:15.000000000","uploader":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"ref":"refs/changes/11/852611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/852611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/852611/1"}}},"commit":{"parents":[{"commit":"f40ac40c2c40d57637312959b680121a9c89949a","subject":"Mock getaddrinfo in cinder store tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f40ac40c2c40d57637312959b680121a9c89949a"}]}],"author":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-10 07:00:01.000000000","tz":480},"committer":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-10 07:00:23.000000000","tz":480},"subject":"remove expired notes for Multi Store feature","message":"remove expired notes for Multi Store feature\n\nChange-Id: Ifff797387ec06e8a1c0da13ebd722df0b993d672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c84197fc8ad21f546db6675a50198e5b3a1b51bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c84197fc8ad21f546db6675a50198e5b3a1b51bb"}]},"branch":"refs/heads/master"},"867263769414890e6129dbce8fa2bc57bb632361":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-08-10 07:11:43.000000000","uploader":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"ref":"refs/changes/11/852611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/852611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/852611/2"}}},"commit":{"parents":[{"commit":"f40ac40c2c40d57637312959b680121a9c89949a","subject":"Mock getaddrinfo in cinder store tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f40ac40c2c40d57637312959b680121a9c89949a"}]}],"author":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-10 07:00:01.000000000","tz":480},"committer":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-10 07:11:43.000000000","tz":0},"subject":"remove expired notes for Multi Store feature","message":"remove expired notes for Multi Store feature\n\nThe Multi Store feature is stable for a long time, need delete the EXPERIMENTAL notes in api guide.\n\nChange-Id: Ifff797387ec06e8a1c0da13ebd722df0b993d672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/867263769414890e6129dbce8fa2bc57bb632361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/867263769414890e6129dbce8fa2bc57bb632361"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6de576fda34f05b9304fc7eafa3c48524909077d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-08-10 07:12:11.000000000","uploader":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"ref":"refs/changes/11/852611/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/852611/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/852611/3"}}},"commit":{"parents":[{"commit":"f40ac40c2c40d57637312959b680121a9c89949a","subject":"Mock getaddrinfo in cinder store tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f40ac40c2c40d57637312959b680121a9c89949a"}]}],"author":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-10 07:00:01.000000000","tz":480},"committer":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-10 07:12:11.000000000","tz":0},"subject":"remove expired notes for Multi Store feature","message":"remove expired notes for Multi Store feature\n\nThe Multi Store feature is stable for server versions, need delete the EXPERIMENTAL notes in api guide.\n\nChange-Id: Ifff797387ec06e8a1c0da13ebd722df0b993d672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6de576fda34f05b9304fc7eafa3c48524909077d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6de576fda34f05b9304fc7eafa3c48524909077d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cb60c1d2facde2f318ec30c8b9cf63844eaccf92":{"kind":"REWORK","_number":4,"created":"2022-08-10 07:19:48.000000000","uploader":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"ref":"refs/changes/11/852611/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/852611/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/852611/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/852611/4"}}},"commit":{"parents":[{"commit":"f40ac40c2c40d57637312959b680121a9c89949a","subject":"Mock getaddrinfo in cinder store tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f40ac40c2c40d57637312959b680121a9c89949a"}]}],"author":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-10 07:00:01.000000000","tz":480},"committer":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-10 07:17:47.000000000","tz":480},"subject":"remove expired notes for Multi Store feature","message":"remove expired notes for Multi Store feature\n\nThe Multi Store feature is stable for server versions, need delete the EXPERIMENTAL notes in api guide.\n\nChange-Id: Ifff797387ec06e8a1c0da13ebd722df0b993d672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb60c1d2facde2f318ec30c8b9cf63844eaccf92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb60c1d2facde2f318ec30c8b9cf63844eaccf92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
