)]}'
{"id":"openstack%2Fglance~852719","triplet_id":"openstack%2Fglance~master~I42dd792323ecb67864bfbd887b96b6cf0057632a","project":"openstack/glance","branch":"master","topic":"plugin-loading-error","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-05-29 06:21:05.000000000","reason":"Change was abandoned"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2022-08-17 16:38:00.000000000","reason":"\u003cGERRIT_ACCOUNT_5202\u003e replied on the change","reason_account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-05-29 06:21:05.000000000","reason":"Change was abandoned"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-05-29 06:21:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I42dd792323ecb67864bfbd887b96b6cf0057632a","subject":"Handle missing internal plugin scenario","status":"ABANDONED","created":"2022-08-10 09:36:07.000000000","updated":"2024-05-29 06:21:05.000000000","total_comment_count":27,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"b90d9f5bd5f2c2da2059a4e7117b4100a200864f","_number":852719,"virtual_id_number":852719,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-29 06:43:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-1,"date":"2022-08-30 13:32:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-08-29 11:12:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":-1,"date":"2022-08-30 13:36:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":-1,"date":"2022-09-01 06:05:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-08-29 11:12:01.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-10 11:10:19.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-08-10 11:17:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-10 15:35:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2022-08-30 13:32:52.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-08-30 13:36:33.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-09-13 07:14:50.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"}],"messages":[{"id":"6b8c0b010472b8177b6f116f0bf7e311217f53f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-10 09:36:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bec9156cbc5ccd04ff8c9cb2bec8d86c2780ba34","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-10 11:10:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c9d309d0126547e66f7a720f23429577eaf75bd4","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-10 11:14:26.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4662e31835bced96b58d4cf63a8805502cbd32f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 11:17:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d4027a84ac4498992ac000d9a5d2d32\n\n- grenade https://zuul.opendev.org/t/openstack/build/6a5d0d9d7aeb4b48a715ab75e5df3278 : SUCCESS in 54m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd46ec2573b640fdb3c8c263cb45aa34 : SUCCESS in 1h 25m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1e37c5754ff488b93a3d12e8ae7fb36 : SUCCESS in 47m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b0b943cb2cf4665a4b15c8f9f3ea411 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/480c0aee7c064277a6f683bab03f14fe : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/099d02170fde499ab1eba2fe0feaaac0 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4303367449f4ff1be9a6f13b7d1b015 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a24edaab447849399f70999d419b2934 : SUCCESS in 7m 09s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/83b55b1988dd43f6b2a0e8562abea32a : SUCCESS in 10m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/91e278f6e6774456ae1f8b8053020fa8 : SUCCESS in 9m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2d6c5e086a17428e81e8a685b2318c99 : SUCCESS in 9m 26s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fdc78acee68546a89756ae430f7ded57 : SUCCESS in 3m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/38160194012949c7afb7c4dcf025d705 : SUCCESS in 1h 17m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/acc395c820d44484bec5b3a11c44dc77 : SUCCESS in 1h 21m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9c4b45a2ac91422280dfc213f1e10598 : SUCCESS in 1h 36m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0d31dbb395ca4931ac49333cea276529 : SUCCESS in 1h 20m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f2077d161ea4329823973ff9e77b655 : SUCCESS in 57m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3457a065a0824c498d3a1e3505578779 : SUCCESS in 1h 05m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b28e6294ae2c4ed395b32c3405e55bd9 : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":1},{"id":"73f1dfea833b2fa39d7c440b9d48993cf8ecd2b1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-10 14:44:45.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"300e86a7bc93def594e0a582aa7412f8a5b4def0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-08-10 15:35:07.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7488068c57fec1959fb4c401aa38bcba4ff6397d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-11 05:10:10.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ab7638c8b722e5391e3a21e20eb5e9cf7ef39822","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-11 05:16:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b25eec6dd1b021c8d0d58451df332058c59d37c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 06:47:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e7b5fa083354811b780fb6756112a38\n\n- grenade https://zuul.opendev.org/t/openstack/build/6aa58b8c9b6c4553bcb408989728e1df : SUCCESS in 58m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f1020b3ac66d4ff3b15c40cf86d9a7e0 : SUCCESS in 1h 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3823b25325b84d0698c3faa25d75271f : SUCCESS in 50m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4dfe482e64848e1bc3218f0f1e2ae36 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dbaa5587a12457fa6708da45caa3db3 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eaf35f558c85496782c22c0b9fefe194 : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f7bcc8fba414056801e5f14e5c366fb : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f95671f3f2e340c6b1839ed5c067cce9 : SUCCESS in 7m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1517769be85d45b3bce620e3b3907b52 : SUCCESS in 10m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0883d1f9ec934af59cc02863f0816e65 : SUCCESS in 9m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/83ef7f22912c4569892a906d4cad285f : SUCCESS in 9m 43s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/04f05c2dd5f54936bcad2be1259c2d78 : SUCCESS in 5m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5b0809a1b35d4903bfa1f79ec1bb308c : SUCCESS in 1h 21m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4e87d56d14064f668ea00df32a3e3a98 : SUCCESS in 1h 10m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7f389ec9d38b4ca4b80f9bf19d9d2200 : SUCCESS in 1h 24m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7668eb6453c64364ad65fe8c8a534b93 : SUCCESS in 1h 29m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9410ffebdaf94f8bbdcecf4d3196880e : SUCCESS in 51m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0f6b6efb0dab45d48ff33db9a82d82bd : FAILURE in 1h 18m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/815304cad10843a89bc3b30902557b6f : SUCCESS in 35m 52s","accounts_in_message":[],"_revision_number":2},{"id":"a2721c505ba7a414c0dec4a2552acf866bfd2a1f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-11 06:51:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8920d561e983558dec1930ab07967f9551fa406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 08:34:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/708a09e174fb47b1bcb9fbefb261954f\n\n- grenade https://zuul.opendev.org/t/openstack/build/3060ca9ec558438c993fffbd9361b3e1 : SUCCESS in 1h 08m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9565715a45c9460fb3ded7f0826a7f11 : SUCCESS in 1h 29m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/343143e439a345b6925f41307c0ac1e4 : SUCCESS in 56m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a89de9a5f17d4b9b8145dc8ad21d34f4 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f7dc5da4be443fab5ee8a5a038cc0b6 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4de8f9df57b4c4da7a969ed003a4248 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55dae44b787b472f8ce724caa55a1b11 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8aadcc8aa0843d5bf1715c3a664ec4e : SUCCESS in 6m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d2ce82d54b3c4797a718259279d0e838 : SUCCESS in 11m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d3565879a01942d49bb3a34a86275765 : SUCCESS in 8m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d42b821513674c81ae23cea143c59f4c : SUCCESS in 9m 42s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/feadb0273d634a81a83eb66ee2f1aebb : SUCCESS in 4m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cf7001e6a9fb4b31b2c05a5cf526fe3a : SUCCESS in 1h 23m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/406f6629c9904a9893d0801609f0475d : SUCCESS in 1h 20m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3cd27af8794e4e5bbc8cd52069759352 : SUCCESS in 1h 25m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/797ba90fdbe1424890f4f0f7d5ad6432 : SUCCESS in 1h 36m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c871a0ff2944e01a9bf5381a2df5090 : SUCCESS in 52m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/764e580284d641deaf53838d3d085d15 : SUCCESS in 1h 10m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7413b61f5b1941fdb95e67dc6e41f5f6 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":2},{"id":"d841fcd530931bb49a549eee045f7d5b68fd53cf","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-17 16:38:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c86f2441e3624fc78b85e35adaa39336c271d38","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-26 12:18:53.000000000","message":"Patch Set 2: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a0b84f65f3770a63fbd5e531a2a6ca651625455c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-29 05:07:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c16537551b304fc6ee9fd837353357bea084bd63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 06:43:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a4a5534768445efba34b0498636e4c2\n\n- grenade https://zuul.opendev.org/t/openstack/build/24d0de0d92ca476dba7a73364c741e96 : SUCCESS in 1h 03m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/caa053c8618c4e94a651e909325e0c53 : SUCCESS in 1h 23m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba3f415112a44bd7865f136143ccaba1 : SUCCESS in 40m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2111188c91c4c11b55cc96507df68c3 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98278163fad14be7b7102e3b2e54a44e : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93628865cfdb411f87eafb7b3f30841f : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2d98a1c8e4e40fda79aea88d8c9308c : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc83616c54844a7e9c18802ddbe87764 : SUCCESS in 6m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c67a5cd31be54e4f825daab1cf2a8c29 : SUCCESS in 10m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f8be743ac554fd6a374630cf0c83230 : SUCCESS in 9m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b2a6605d5f814967b3c22d0e81b37d8b : SUCCESS in 9m 15s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4b823203975e4cd9a93246a3238f0502 : SUCCESS in 5m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7b6788456a6e4b4d88f1e1f202165f3e : SUCCESS in 1h 13m 37s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8513cbe1ecce46269490e60354a17d22 : SUCCESS in 1h 10m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/18b89b4bedd74513bf5346fea28b4f22 : SUCCESS in 1h 17m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2be27cc12f9d407db04db575f1d719a4 : SUCCESS in 1h 34m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2d242a2d8ff42eb9a69d64104a970bc : SUCCESS in 45m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e29214465371496b9744472affbe9663 : SUCCESS in 1h 12m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d7a185a07cbc470d9c015bf34bb47667 : SUCCESS in 43m 22s","accounts_in_message":[],"_revision_number":3},{"id":"bdad76efd075b54f9cd157df293f1892e0ee2846","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-29 11:12:01.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"44ae48905c3755b6e232b15b6126bdba79b44e68","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-30 13:32:52.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d218b5b89ade47d5f252c23ac71868a6f3550fee","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-08-30 13:36:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4c03a4d252bcf7e206faf56b04041a2ee75be65f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-09-01 06:05:42.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"78abc2560986608a3725443b9f5e01b077f80279","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-09-13 07:14:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b90d9f5bd5f2c2da2059a4e7117b4100a200864f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-29 06:21:05.000000000","message":"Abandoned\n\nWe will revisit this in next PTG!","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3102a7f5479704c6237122e43f976a57f45200a1","revisions":{"cead9805848c379c254319b553ab2d928c271a81":{"kind":"REWORK","_number":1,"created":"2022-08-10 09:36:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/852719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/852719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/852719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/852719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/852719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/852719/1"}}},"commit":{"parents":[{"commit":"f40ac40c2c40d57637312959b680121a9c89949a","subject":"Mock getaddrinfo in cinder store tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f40ac40c2c40d57637312959b680121a9c89949a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-08-10 08:59:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-08-10 09:30:35.000000000","tz":0},"subject":"Handle missing internal plugin scenario","message":"Handle missing internal plugin scenario\n\nIf internal plugin is not expicitly mentioned in the setup.cfg\nthen loading of the plugin fails at the time of import API with\nstacktrace by which it is difficult to understand the exact cause\nof failure.\n\nAdded exception with useful error message for the same as well as\nreleasing the task lock so that the same image can be reused for\nfurther import actions.\n\nChange-Id: I42dd792323ecb67864bfbd887b96b6cf0057632a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cead9805848c379c254319b553ab2d928c271a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cead9805848c379c254319b553ab2d928c271a81"}]},"branch":"refs/heads/master"},"923683bad330a948e3be985c06c82a40a4576977":{"kind":"REWORK","_number":2,"created":"2022-08-11 05:16:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/852719/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/852719/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/852719/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/852719/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/852719/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/852719/2"}}},"commit":{"parents":[{"commit":"f40ac40c2c40d57637312959b680121a9c89949a","subject":"Mock getaddrinfo in cinder store tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f40ac40c2c40d57637312959b680121a9c89949a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-08-10 08:59:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-08-11 05:15:00.000000000","tz":0},"subject":"Handle missing internal plugin scenario","message":"Handle missing internal plugin scenario\n\nIf internal plugin is not expicitly mentioned in the setup.cfg\nthen loading of the plugin fails at the time of import API with\nstacktrace by which it is difficult to understand the exact cause\nof failure.\n\nAdded exception with useful error message for the same as well as\nreleasing the task lock so that the same image can be re-used for\nfurther import actions.\n\nChange-Id: I42dd792323ecb67864bfbd887b96b6cf0057632a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/923683bad330a948e3be985c06c82a40a4576977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/923683bad330a948e3be985c06c82a40a4576977"}]},"branch":"refs/heads/master"},"3102a7f5479704c6237122e43f976a57f45200a1":{"kind":"REWORK","_number":3,"created":"2022-08-29 05:07:45.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/852719/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/852719/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/852719/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/852719/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/852719/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/852719/3"}}},"commit":{"parents":[{"commit":"2d1b460993197a6691488349114b0be7543fea33","subject":"Merge \"Add documentation for glance-download plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d1b460993197a6691488349114b0be7543fea33"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-08-10 08:59:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-08-29 04:59:06.000000000","tz":0},"subject":"Handle missing internal plugin scenario","message":"Handle missing internal plugin scenario\n\nIf internal plugin is not expicitly mentioned in the setup.cfg\nthen loading of the plugin fails at the time of import API with\nstacktrace by which it is difficult to understand the exact cause\nof failure.\n\nAdded exception with useful error message for the same as well as\nreleasing the task lock so that the same image can be re-used for\nfurther import actions.\n\nChange-Id: I42dd792323ecb67864bfbd887b96b6cf0057632a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3102a7f5479704c6237122e43f976a57f45200a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3102a7f5479704c6237122e43f976a57f45200a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
