)]}'
{"id":"openstack%2Fglance~853927","triplet_id":"openstack%2Fglance~master~Ib4f19351ca7985474d838998b402d3c5f9fb195f","project":"openstack/glance","branch":"master","topic":"bp/glance-download-import","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-08-23 17:16:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"Ib4f19351ca7985474d838998b402d3c5f9fb195f","subject":"Add support to get glance endpoint","status":"MERGED","created":"2022-08-22 08:02:39.000000000","updated":"2022-08-23 20:28:29.000000000","submitted":"2022-08-23 20:27:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853927-bp/glance-download-import","meta_rev_id":"a36449a78e584e5d6c7da2c3f200be7cc3b6970a","_number":853927,"virtual_id_number":853927,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 20:27:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-08-23 15:39:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-08-22 14:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-08-23 15:39:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 09:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 14:05:29.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"b3eb8d83125609513a79d4b70a81fab87f9330fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-22 08:02:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95bd5e54e355fa3cf6b1467b30feb8289eb9663e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 09:46:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f63f50c1dba44c58825bfb79b7edcfa6\n\n- grenade https://zuul.opendev.org/t/openstack/build/e10d697921f34497a02db7b15b58c73e : SUCCESS in 47m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87e065466b2f436682b235cd6ba751e1 : SUCCESS in 1h 33m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3af5782a58504384b24ace7203564949 : SUCCESS in 46m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618590b7939a4429b5a0479bfeb9ff9b : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ff2e4cd81ce43d29cb7df6dc61ce4aa : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/372637a1389945dfb8c65cc73e4a813b : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a0b7e95feab4b81afc125e18d21b62f : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f87f5b4f634c68a92b27cab09f5b16 : SUCCESS in 7m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7fe8b24edf7648e4b3a5fc45dd8078de : SUCCESS in 11m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f1dc99c977fd4383a8c9b2f49d84f495 : SUCCESS in 11m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dcf97e37bf394e039e63a2dfe3ee15ae : SUCCESS in 8m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/09750cfeaef6482e8069d6668ddbd822 : SUCCESS in 6m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a45432442aaf4ba29fd7833bcca63236 : FAILURE in 1h 13m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/54fb974d32614e05b68d535ad510ed54 : SUCCESS in 1h 35m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b5816db5dfa64cf2882435b240cf32a3 : SUCCESS in 1h 12m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f545535d9ad14e4db6b4b2b89de38f33 : SUCCESS in 1h 41m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02a65246bcde48e69a8b4a0093d3303d : SUCCESS in 47m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d2ef498d480b4454af0b6acba5f64a1c : SUCCESS in 1h 08m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4f332d186b194c6cbfd6bf1f45e5d576 : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4c18d4fccc8ff93a5a6d2fb4e183a4427b2abfae","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-08-22 14:05:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f6fb74a1cb9b64fc95ce32af83648c9eb3970cb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-23 15:39:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85b2494d7b38d31c52e42c7884ab94b21c0c07c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 15:40:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"707a4396db5f6ed4595275aa30a79f25646a2733","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 17:10:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03ef6ceeda5f43a49358a278dad25057\n\n- grenade https://zuul.opendev.org/t/openstack/build/7b4a2cb2cc2e4781820fe4bf074be905 : SUCCESS in 1h 08m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cdd007754f894bada59b61c6bcb2a7c4 : SUCCESS in 1h 29m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f3ff263777a4f708f27555ce4b91b5c : SUCCESS in 58m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c320c0d66fac4b08873e291938bdd8b3 : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d31f00ba14a145ddb9a80143c2b8645d : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2a0dff678a047e0b2fadaf2267dcc66 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18331b52902647a8b9b2523e8fa1897d : SUCCESS in 8m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/26ca9c36b5584fa58fd5ea87b55c2466 : SUCCESS in 12m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d882e8608bbd4958a68e8f58c07a1400 : TIMED_OUT in 31m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dbea37b2427d45a8b870eaa9aaa8c9fc : SUCCESS in 1h 24m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fee5dc189298424b8c5c0964d9b3d68a : SUCCESS in 1h 28m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69d5d29bf2254efab7941d85ef506706 : SUCCESS in 59m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/83267ba3b08f4f0888abe0279266293c : SUCCESS in 1h 27m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/31c1d1fcdf8a4ff192ccd89f34e1e5f0 : SUCCESS in 43m 46s","accounts_in_message":[],"_revision_number":1},{"id":"eac0bc32d9cf7ffcb6fac28019e1ec0b8fcc3d3a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-23 17:16:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6af6c4e8d682d5bfe36028e3b18062ef4ed35979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 18:50:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e03c274a8af54fa9abd1a33e677e562e\n\n- grenade https://zuul.opendev.org/t/openstack/build/c16dea1dd886449d9b5fd650e4c7b97c : SUCCESS in 57m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/14e0ac88146b4e599049f016bc0d6bbc : SUCCESS in 1h 20m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cffadb062de94d35848897d839486f21 : SUCCESS in 51m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2a7398a076a49969486770b6c9dcc8b : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/626da65f7c1246caa50b64c4316578b5 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f073620ef4514af5be60d200286caf50 : SUCCESS in 9m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4681aa3221154fc0800d33eac53e2ed5 : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418fa37580774e858a61eec2de7df0cd : SUCCESS in 8m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ef40ce6ae575428db343a3b484765bab : SUCCESS in 14m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c7bb75965dea463095654bf4882ed055 : SUCCESS in 11m 14s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9226e2bcaec7477c8319bc4778807704 : SUCCESS in 13m 09s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f8d4623c44e84ca3b9b5fc61ff8f0245 : SUCCESS in 8m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/95943e795d974b86aa94cf82ba0f7061 : SUCCESS in 1h 11m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0015787526d9406fb9eb4f5581952e20 : SUCCESS in 1h 12m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/da30a93d6e5242919d8e44a45d676574 : SUCCESS in 1h 31m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/036af3ad0c544434a98ed852e6c63bb2 : SUCCESS in 1h 27m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6958b9e5a44246b481a6959070c22cea : SUCCESS in 1h 00m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aa9c998dc80542fa814fdf6efd08a497 : SUCCESS in 1h 07m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e400e17998cc4973b922e97682dcbaca : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d9380fce84aeab39ea72255340dda1434983ed5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 18:50:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15747c7b82eb7ff18880811f00b1078357b17036","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 20:27:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0d8b366b596416c9bbbf9b84c846539\n\n- grenade https://zuul.opendev.org/t/openstack/build/bc9ba403fd2a4912b500e1b66b731287 : SUCCESS in 55m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/053209d6c77e439f86f1316b94e2d3b6 : SUCCESS in 1h 18m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9845f5205ec2410baf2a119e70148aec : SUCCESS in 48m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e58182dae1d4ee2887f041b6c094356 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91597db8ba274c928f605a40917b51e2 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bc637c4a9b14a99863a6d1f5e6834b0 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ead106a40f4e45a0aa5bc517c023f213 : SUCCESS in 7m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf84300b06144509b34b7b13f4fd37b4 : SUCCESS in 10m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/138ebf32e85d4e3a8dc5d2e063ee9ac0 : SUCCESS in 13m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d0e8ec56a2574f8d8b9f87ab5dbff9ff : SUCCESS in 1h 34m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9fb9c35805c547ceb7aad555fadf88d1 : SUCCESS in 1h 33m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b390e9baaaf43e4ad910efa388bd5ed : SUCCESS in 53m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/291fd4f887d547c098f9cb5d05d60afc : SUCCESS in 1h 06m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8d58ffa79ebc4b3ca64d3f92dc25c20c : SUCCESS in 40m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5806ff1c383ad5c578dd25e63623be25b46f8f91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 20:27:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a36449a78e584e5d6c7da2c3f200be7cc3b6970a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 20:28:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40b4f967d2cd4dcab623acedc6e4d3e4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/296ea93290864958a23eb7ebf62813a0 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"653d52e90dbba5284f315251099275bac5eb1a2b","revisions":{"653d52e90dbba5284f315251099275bac5eb1a2b":{"kind":"REWORK","_number":1,"created":"2022-08-22 08:02:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/27/853927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/853927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/853927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/853927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/853927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/853927/1"}}},"commit":{"parents":[{"commit":"cb60c1d2facde2f318ec30c8b9cf63844eaccf92","subject":"remove expired notes for Multi Store feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb60c1d2facde2f318ec30c8b9cf63844eaccf92"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-08-22 06:44:47.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-08-22 06:44:47.000000000","tz":0},"subject":"Add support to get glance endpoint","message":"Add support to get glance endpoint\n\nThis patch will help to get glance endpoint form service\ncatalog based on input task.\n\nPartially implements: blueprint glance-download-import\nCo-Authored-By: Victor Coutellier \u003cvictor.coutellier@gmail.com\u003e\nCo-Authored-By: Pierre-Samuel Le Stang \u003cpierre-samuel.le-stang@corp.ovh.com\u003e\n\nChange-Id: Ib4f19351ca7985474d838998b402d3c5f9fb195f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/653d52e90dbba5284f315251099275bac5eb1a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/653d52e90dbba5284f315251099275bac5eb1a2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
