)]}'
{"id":"openstack%2Fglance~854495","triplet_id":"openstack%2Fglance~master~I5febc9f4094f88caab3ed18026a8cb7b6a345d45","project":"openstack/glance","branch":"master","topic":"zed-m3-config-refresh","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2022-08-30 07:01:15.000000000","reason":"Change was submitted"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2022-08-29 11:10:36.000000000","reason":"\u003cGERRIT_ACCOUNT_5202\u003e replied on the change","reason_account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-08-30 05:13:49.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I5febc9f4094f88caab3ed18026a8cb7b6a345d45","subject":"Refresh Glance example configs for zed milestone 3","status":"MERGED","created":"2022-08-25 04:49:42.000000000","updated":"2022-08-30 07:02:19.000000000","submitted":"2022-08-30 07:01:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854495-zed-m3-config-refresh","meta_rev_id":"18b095ec214c0b4e87deb4edb5979d60ae463cb4","_number":854495,"virtual_id_number":854495,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-30 07:01:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2022-08-29 11:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-08-30 05:13:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-08-30 05:13:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-25 05:38:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-08-25 05:56:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 03:00:03.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-08-26 11:02:12.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"b3fd0b32fb92dd6eed570d9f71692047834596bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-08-25 04:49:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9846e275232242e375c20f0838aea2345c1b3e33","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-25 05:38:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"673fe24bc6d48cb12caa64d6de691c74ebd161a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 05:56:42.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"7c9bf316a271c8f47cb982c44f369f61b0f485c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-25 06:00:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6cc9ce643b27ab227480e6c7fb6d8bf1a967603c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-25 06:00:51.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6ddc2319963fae8edc7ad50404858bf3701563e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 07:42:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c80e060f12a43a3998013b6fe01cc08\n\n- grenade https://zuul.opendev.org/t/openstack/build/2ebd4f557acf41448571f6d76537fb11 : SUCCESS in 55m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c258dc0e4ff4797b62ff17b42f240e8 : SUCCESS in 1h 12m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/129b922275cb4ab491b4c61591edb6d4 : SUCCESS in 48m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/279b6235acb04eb0840ff59ea6358fa0 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0cc94af529041ea876180a06ba8ccd7 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6192819d9bc645b7b955cf647395a82a : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad15e6df8b774df49c2f928afbc1a357 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce066f510613425494466ad7c730c232 : SUCCESS in 6m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4a2530615a2a4bd78902fde66a4ecf4f : SUCCESS in 11m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/81cadc05560c405baa22b72f9d8745f8 : SUCCESS in 9m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b94e9b4fcfc745caa39d4f631419c8f3 : SUCCESS in 7m 56s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3a64559d6c2b47eba3039940acc5dc28 : SUCCESS in 1h 13m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5f9a2c9b60724733b6d979508093ca9f : SUCCESS in 1h 15m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1a4376bac86c4dea8a3ee25477cb6a11 : SUCCESS in 1h 10m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c42ae729eb4047c4ba7e47c0ac67604c : SUCCESS in 1h 37m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53ce54cdc2f24007be2d752a4845802c : SUCCESS in 44m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/70c1a2c06d8f4b698ffa8f66971aa33e : SUCCESS in 1h 08m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a31b732aaf3a472cb0b3c2c59b47075b : SUCCESS in 31m 55s","accounts_in_message":[],"_revision_number":2},{"id":"9da774b3577c532e14bede6660a1348e0572d4bb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-26 03:00:03.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5a5acda15edcc8198e474a6bdc33dfe19f99a125","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-26 05:53:55.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f4e06947372b6a81d9268b4b9d06e13a54445721","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:00:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54980f792fcc402481301d1e72a17bc8\n\n- grenade https://zuul.opendev.org/t/openstack/build/a9007084f4d746698bdbbdc1468d9246 : SUCCESS in 55m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2dbbafaa5a494e8a97684381210a0530 : SUCCESS in 1h 18m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96a27c32de084887b148f6abc73f28a4 : SUCCESS in 57m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/181adc4312d24fb591ea050a61bf0610 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cd6ed12c07545619cbf9f86ce775d33 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7feba66cf3247268c6b6b9c2b0997bd : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79a9a2450acc4cedb9e6eda44c85e58b : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6c027531c724a8088d93f7e76256f0b : SUCCESS in 7m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/316430e8cc1d40f2b8037fbaff8bc499 : SUCCESS in 11m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c7166842b294baebed82838180c60b3 : SUCCESS in 10m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a496bae7ae9f4e6792b9f9d0976d8932 : SUCCESS in 10m 32s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6c029dc334c7451d85dda96e57ac498b : SUCCESS in 1h 18m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/46b3416c2830455e871404e78bce31d1 : SUCCESS in 1h 20m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bee14738300443dba7da349824570be4 : SUCCESS in 1h 18m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ed29d431bdbd4a1186ff6b14275e1741 : FAILURE in 2h 03m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e08577035074410db6a6880354baea25 : SUCCESS in 53m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b072b331fb364289904015c43634341a : SUCCESS in 1h 48m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a9b5c0d6a7f342279bc3d97352ce5805 : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":3},{"id":"46ac606f81fb71dacad69535b43436bdebbc4820","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-26 08:58:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9e14cd02e3f41fcbba2a29b7e292101286d48b21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 10:52:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b65f6aa9264810afe398ba1a5edb02\n\n- grenade https://zuul.opendev.org/t/openstack/build/be913e7bdfaa43b195d48acd8ca3d2ca : SUCCESS in 1h 06m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f30573eeaab464986744bf59b125c68 : SUCCESS in 1h 21m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa09df1a2a70472baec79ba20c55e654 : SUCCESS in 1h 02m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d69aec5b90a84884aed2e4b01f358b8e : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/201aecde10804687864766c689c0254d : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ee5f134d7364caa869bcf045f79cc20 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c11a0a26d73e41ed89e97923d326c58a : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ae5c38287547c9bb361eb694a1d725 : SUCCESS in 6m 25s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/20b0f6c8bdee4bc59d8557697673bb5e : SUCCESS in 12m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce88a6d983184911a4b81d55f809c406 : SUCCESS in 9m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/217cf7796d984bdfbbc5ca5921297d15 : SUCCESS in 8m 59s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/05f0a48cdf7d46df944eaa4c739edf1d : SUCCESS in 1h 19m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/820b61917c024d6fa4862319a27553d2 : SUCCESS in 1h 39m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aa7e910160954420bc819a667f62fbdb : SUCCESS in 1h 27m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/baa69a849504451792e4d7b7a435685f : SUCCESS in 1h 46m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f10fddc942041268cace95c79bb17d1 : SUCCESS in 1h 00m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6c60f54fe5714688a18f4bd247dd46e1 : SUCCESS in 1h 16m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/46a24311b7614a1da7a5a628ff9fb3ce : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":3},{"id":"459188268815d9ab7370d73254f3364788bdcf04","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-26 11:02:12.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0db99625b2dea9df4fb3c49a96f1608bfaeda03f","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-26 11:53:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"821464c60ba7bb9ca68b5c23eeca69e3ef850fc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-08-29 07:34:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"83a1e36e8067f576de631a1f588d7757d682a309","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-08-29 07:35:26.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5c189b39f39da7825f7f2f67860dc7d39092371c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 09:13:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e7bfa4a260d42b78401390353e6ac0c\n\n- grenade https://zuul.opendev.org/t/openstack/build/cd117d291657429189092790c6996ed0 : SUCCESS in 52m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/441ec9e4544e41eda8beed544b7c2675 : SUCCESS in 1h 18m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76c1e8a690334b0cb4fa999231447935 : SUCCESS in 47m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a66cb6cea3a41bda1a8b4901c47d4d6 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d20d1ac3aed474e9a02b2694013b1fc : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6b35929c9d74e20ac85ca1c0b01eb36 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/008d8f10d27843598f2e87708a2c6478 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/000d918dd6804078899040856f3cdaea : SUCCESS in 6m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/06b5d9465390452d811a8b1368a0a4dd : SUCCESS in 12m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c553d92746cc4345909c5bb22fcf338c : SUCCESS in 10m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d5af06aef3c041eb83b136f543708c2d : SUCCESS in 9m 55s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0655c1e8bed74271aa18256331607873 : FAILURE in 1h 05m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/742fb1cfae454e8e8e17b39a6fe97a7f : SUCCESS in 1h 27m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/04b680a2cf204069a69a8172159dcd26 : SUCCESS in 1h 20m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0bce0fa10cff49fbaf7f9ecf4d3058e5 : SUCCESS in 1h 37m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f28f71421aa4319804c404692cfb47f : SUCCESS in 50m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dfdcf58e28074cc9a50f3b54b885e0be : SUCCESS in 1h 09m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cb0188d9c87d4fc59aa2691a76ac305b : SUCCESS in 36m 21s","accounts_in_message":[],"_revision_number":4},{"id":"59cddeef725dd38376254368425424c73a5dee90","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-29 11:10:36.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"681a231fee946b04d90c04942d8d8da183217fe4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-30 05:13:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7efa2a83da862f9155f781a0aec40636005f9f9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 05:14:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a7f6157a4a3b5f1e69544ec2a9d558d614a75086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 07:01:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f72bd2e6c7143988556715faf7d0bba\n\n- grenade https://zuul.opendev.org/t/openstack/build/f2a037661c2848f9b419a035c7bfdf44 : SUCCESS in 55m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d66144a91db641e6bea6e0d53ecbf6cb : SUCCESS in 1h 45m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/447cd2b3c22e455c93bf4922737947df : SUCCESS in 51m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cd71e0479f646ba8537b802f8050b4c : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd142b36218a4c75a02e28d1bfddf79d : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06fff90285fc4fa5a3fbef9d283f8ba0 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5204904673b14a10840cf9e1789eba25 : SUCCESS in 8m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/50b8d2484eb64f7a8dfd7cbd4ac254e7 : SUCCESS in 8m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/96ed7378a0c9499e946643f4d16a6f1f : SUCCESS in 11m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1ba2297c11f542d18f6afe712f6509c4 : SUCCESS in 1h 31m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5cf6a8aac7dc4a82844a77d14c3599fe : SUCCESS in 1h 36m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b22c3fc553943699bb74c5ac0cd6bed : SUCCESS in 1h 08m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9bfc1d5191934df484ffec6f1d132a00 : SUCCESS in 1h 13m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/619b9a72b39d496a9a30508dcaab296a : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":4},{"id":"9ef04c0999c530852399455693f10aee42ec2e07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 07:01:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"18b095ec214c0b4e87deb4edb5979d60ae463cb4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 07:02:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcf9c8fcf1014e2082d13c741fdda646\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a1e8e3c79746a6b08f8fa06efad709 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5d38dbdb9c9e306a66b5435c89c38e732fb118cd","revisions":{"2f34a012cd8a957d4bffa3d92693c821ce369d47":{"kind":"REWORK","_number":1,"created":"2022-08-25 04:49:42.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/95/854495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/854495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/854495/1"}}},"commit":{"parents":[{"commit":"5c675bddcf7cc1af5b5cab3186a5a76ffde76a1c","subject":"Merge \"Extending stores-detail API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c675bddcf7cc1af5b5cab3186a5a76ffde76a1c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-08-24 15:18:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-08-25 04:47:01.000000000","tz":0},"subject":"Refresh Glance example configs for zed milestone 3","message":"Refresh Glance example configs for zed milestone 3\n\nDepends-On: Ie36232f3474d398a913025c8d408f08f879874d8\nChange-Id: I5febc9f4094f88caab3ed18026a8cb7b6a345d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f34a012cd8a957d4bffa3d92693c821ce369d47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f34a012cd8a957d4bffa3d92693c821ce369d47"}]},"branch":"refs/heads/master"},"8a92dcaa79b522b1f61d08da2f3200b3d32a3142":{"kind":"REWORK","_number":2,"created":"2022-08-25 06:00:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/95/854495/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/854495/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/854495/2"}}},"commit":{"parents":[{"commit":"628d0a304f38cfc8a0a332e68ad18ceeea44383f","subject":"[glance-download] Add missing config options to config list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/628d0a304f38cfc8a0a332e68ad18ceeea44383f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-08-24 15:18:09.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-08-25 05:57:07.000000000","tz":0},"subject":"Refresh Glance example configs for zed milestone 3","message":"Refresh Glance example configs for zed milestone 3\n\nChange-Id: I5febc9f4094f88caab3ed18026a8cb7b6a345d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a92dcaa79b522b1f61d08da2f3200b3d32a3142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a92dcaa79b522b1f61d08da2f3200b3d32a3142"}]},"branch":"refs/heads/master"},"599cd483e46d902a13af8e96cd3b9fcfa4747cd6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-26 05:53:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/95/854495/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/854495/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/854495/3"}}},"commit":{"parents":[{"commit":"91f8f16548a6889ae56350011b56611f50bf7986","subject":"[glance-download] Add missing config options to config list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91f8f16548a6889ae56350011b56611f50bf7986"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-08-24 15:18:09.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-08-26 05:50:53.000000000","tz":0},"subject":"Refresh Glance example configs for zed milestone 3","message":"Refresh Glance example configs for zed milestone 3\n\nChange-Id: I5febc9f4094f88caab3ed18026a8cb7b6a345d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/599cd483e46d902a13af8e96cd3b9fcfa4747cd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/599cd483e46d902a13af8e96cd3b9fcfa4747cd6"}]},"branch":"refs/heads/master"},"5d38dbdb9c9e306a66b5435c89c38e732fb118cd":{"kind":"REWORK","_number":4,"created":"2022-08-29 07:34:26.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/95/854495/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/854495/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/854495/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/854495/4"}}},"commit":{"parents":[{"commit":"51e4fb1485b7c3c69ea90b3cea901dc446c5cc50","subject":"Merge \"[glance-download] Add missing config options to config list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51e4fb1485b7c3c69ea90b3cea901dc446c5cc50"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-08-24 15:18:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-08-29 07:29:04.000000000","tz":0},"subject":"Refresh Glance example configs for zed milestone 3","message":"Refresh Glance example configs for zed milestone 3\n\nChange-Id: I5febc9f4094f88caab3ed18026a8cb7b6a345d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d38dbdb9c9e306a66b5435c89c38e732fb118cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d38dbdb9c9e306a66b5435c89c38e732fb118cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
