)]}'
{"id":"openstack%2Fglance~854772","triplet_id":"openstack%2Fglance~master~Icc792e23aad9e7a9a788e6b5826bc5aae54ae978","project":"openstack/glance","branch":"master","topic":"fix_default_import_methods","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2022-08-26 19:43:15.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-08-26 12:49:33.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}},"hashtags":[],"change_id":"Icc792e23aad9e7a9a788e6b5826bc5aae54ae978","subject":"Remove \u0027glance-download\u0027 from default import methods","status":"MERGED","created":"2022-08-26 11:30:05.000000000","updated":"2022-08-26 19:45:11.000000000","submitted":"2022-08-26 19:43:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854772-fix_default_import_methods","meta_rev_id":"7c12a91c8f78ddf74eb028cde47ac98c5abfc899","_number":854772,"virtual_id_number":854772,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 19:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-08-26 17:53:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2022-08-26 17:40:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-08-26 17:53:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"date":"2022-08-26 17:53:11.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-08-26 19:45:11.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 12:37:21.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-08-26 14:29:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 15:50:54.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-08-26 17:40:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"31496259a5b89f320cb7d9e60076747e1d5c626e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-26 11:30:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f9f65abc75049823b2841bdfd14e0209584fd54","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-26 12:37:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"40be246d4dbac7afd8de6c9b997903ec73232017","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-26 12:38:31.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"c6f0b0e6e9d2fdc85b8c0ab9aeeae983e15e7674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-26 12:46:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cff8e97460e501f212811995babfe232e2080160","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-26 12:47:18.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5beb68959d5e5ac0d97d76e6eda54c2995ea24e2","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-26 12:48:15.000000000","message":"Patch Set 2: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"695ae321dd1944182a316687a5625dc154f7e3f0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-26 12:49:33.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6a4b08109e7d41724a8dfbbd388c3ea59897cf2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 14:29:17.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0296062c2af461cb8f3b85082e011fa\n\n- grenade https://zuul.opendev.org/t/openstack/build/1b500d508f614fa08b57bf1f267dc52c : SUCCESS in 57m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5682c18984f148638e40b26ffba99eea : SUCCESS in 1h 26m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b32fa47d052427daa55e9d2263f820f : SUCCESS in 46m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/500014f96f2243be99f36984aecb34d5 : FAILURE in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/880dd71d98ab4de396cfbab29f0404b0 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dad87d1b9b964b76a5f478c0f967042f : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/365e89dd69ac433bb2bcd7451b191255 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57660542d40542708bb59fd920ba3aeb : SUCCESS in 8m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d614943cc81a4982872853a506484331 : SUCCESS in 12m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7edf83c4ca524f77bbb6258fcd954b6d : SUCCESS in 8m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a5561b0f73b54823bd7770b2d3f4f044 : SUCCESS in 9m 52s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6f3fd37c478249dc8bc956d10fae1daf : SUCCESS in 4m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b3dfc08bda1e4e48a406f989efa3740b : SUCCESS in 1h 20m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c413721294534428b992fd07f7e6cfa1 : SUCCESS in 1h 33m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b0c21984b5654ebca5331d29a9535aec : SUCCESS in 1h 26m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cd64e506c8554a1ea6724408af0dd30e : SUCCESS in 1h 39m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d6f45c00498473fa5565b79a2c74376 : SUCCESS in 55m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/00bca4e60f2846b0804ef027e14fd16b : SUCCESS in 1h 20m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c463f081829d48ae936d6b232f537335 : SUCCESS in 44m 57s","accounts_in_message":[],"_revision_number":2},{"id":"9e49df6a1d9ae681c2a5a3a865ebd6912408ff64","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-26 15:50:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a94f8d5a0513d987ff8cbcbe93e04a5b5ee14f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-08-26 15:57:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4c98065a9a80dc11214d5e8016cde845281c9c42","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-26 17:40:33.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2cf47231ccec301cdb2d53b9ddc0b10c7df8f4ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 17:41:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6df6c878a6d04c2fb99c059092d193c7\n\n- grenade https://zuul.opendev.org/t/openstack/build/7d96f300878b496ca0fc773dbdd07db9 : SUCCESS in 1h 11m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f19c382880734a2e9a1f9f1bf1d38fb7 : SUCCESS in 1h 38m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f3b365b4c904aa197ecfc5f3865c3fa : SUCCESS in 47m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e5c4504abb04d58bbd6a1362520697f : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e7d0b99998a47dc987b767b5b12200c : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/406009040f3144a1b7c93a2deacc8390 : SUCCESS in 7m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8049ef61da04b30bd6887898b9528e8 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43bf13def7d42c8a6983398432ab04e : SUCCESS in 6m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/12945e0c72244b178784c1f2edcebccf : SUCCESS in 11m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ee28913d90854c70bdb7cdf1d3808d44 : SUCCESS in 9m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d27ea46862e54387becde2a54311d66b : SUCCESS in 9m 40s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ea5adf218bc94acaa89c004864702866 : SUCCESS in 5m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4f3f361f117649499e87d43129f4b614 : SUCCESS in 1h 14m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d41e2dc6333149ea975f7452c404ec77 : SUCCESS in 1h 17m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/065877d098594a43bb67ac37a5e162ad : SUCCESS in 1h 20m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3a25e1b5a1fe48e2bb9eed8b2fb12cb3 : SUCCESS in 1h 36m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9cf693a920b4c21b28a804475ea498f : SUCCESS in 49m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2efa14cec31d401fa9258c31d4e7144f : SUCCESS in 1h 15m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/64a1cc28fbe843cf9f691b95430d97cf : SUCCESS in 33m 32s","accounts_in_message":[],"_revision_number":3},{"id":"01922f465ede6bc01bb536a7f80b14e844333551","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-26 17:53:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1 -Review-Priority\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"87ba2e363d4b2bd4094da9767316dbd577b61c63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 17:53:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"579eaa2123a6c0383848c181ae18eff04c7e928d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 19:43:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73c3fba00028418a915ed9c9cf836e84\n\n- grenade https://zuul.opendev.org/t/openstack/build/a2c7cf414a7446f0aaf8ada16a09c743 : SUCCESS in 55m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f3b115649ae4dc0b92c60aea9973263 : SUCCESS in 1h 26m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a413bf370dd3400d9d9c8421a9cf7c5e : SUCCESS in 50m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/344795f7adc24f52a28dfbcd358efa7f : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d96d6863e0c54ce6b3acc2d476427f07 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5fcffa0cc8049e29beda8dbe9840282 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b14e6775749f4e23a3581cb0bb19b3de : SUCCESS in 7m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/90d572654d94434a99e74ad2afe6e37c : SUCCESS in 9m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5b0040ed693e49b08d9d8bbb9cc236b6 : SUCCESS in 12m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7b40f168df8a4c4885b5ec5dd88e3018 : SUCCESS in 1h 31m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/07a81be92ab84ab387d3800274c01b8d : SUCCESS in 1h 22m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/412a475b6b6c4290a33b94f7250ed629 : SUCCESS in 51m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0ac4887323334d90977cc76176b13be4 : SUCCESS in 1h 14m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7998cf26f46f45efa3136e3bf7bb0198 : SUCCESS in 38m 13s","accounts_in_message":[],"_revision_number":3},{"id":"11369874be4a8f1729c167ecf69410bb0e3cef5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 19:43:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1ce9c85241ba05fe9544f47647dc71407da8d42a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 19:45:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c30445899104546bc82dc2d9940cc81\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f99cabfab6dd4c91bfdabb63ba55833b : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c342c0e944126c6d257fab8910477a58bdec5da7","revisions":{"cc02758c85c5958b09486dded30c260cc512c1bb":{"kind":"REWORK","_number":1,"created":"2022-08-26 11:30:05.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/72/854772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/854772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/854772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/854772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/854772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/854772/1"}}},"commit":{"parents":[{"commit":"623fd75d394e802011aa5908be9d3aa236dbe46f","subject":"Merge \"Fix a deprecation warning about threading.Thread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/623fd75d394e802011aa5908be9d3aa236dbe46f"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-08-26 11:16:06.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-08-26 11:16:06.000000000","tz":60},"subject":"Remove \u0027glance-download\u0027 from default import methods","message":"Remove \u0027glance-download\u0027 from default import methods\n\nAs \u0027glance-download\u0027 requires federated deployments it should\nnot be enabled by default.\n\nAdded line explaining it being available for such deployments\nin the config help text.\n\nChange-Id: Icc792e23aad9e7a9a788e6b5826bc5aae54ae978\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc02758c85c5958b09486dded30c260cc512c1bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc02758c85c5958b09486dded30c260cc512c1bb"}]},"branch":"refs/heads/master"},"400fec0aaa671db19f3a139a0bb2ba038550e379":{"kind":"REWORK","_number":2,"created":"2022-08-26 12:46:00.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/72/854772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/854772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/854772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/854772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/854772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/854772/2"}}},"commit":{"parents":[{"commit":"623fd75d394e802011aa5908be9d3aa236dbe46f","subject":"Merge \"Fix a deprecation warning about threading.Thread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/623fd75d394e802011aa5908be9d3aa236dbe46f"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-08-26 11:16:06.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-08-26 12:45:16.000000000","tz":60},"subject":"Remove \u0027glance-download\u0027 from default import methods","message":"Remove \u0027glance-download\u0027 from default import methods\n\nAs \u0027glance-download\u0027 requires federated deployments it should\nnot be enabled by default.\n\nAdded line explaining it being available for such deployments\nin the config help text.\n\nChange-Id: Icc792e23aad9e7a9a788e6b5826bc5aae54ae978\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/400fec0aaa671db19f3a139a0bb2ba038550e379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/400fec0aaa671db19f3a139a0bb2ba038550e379"}]},"branch":"refs/heads/master"},"c342c0e944126c6d257fab8910477a58bdec5da7":{"kind":"REWORK","_number":3,"created":"2022-08-26 15:57:09.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/72/854772/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/854772/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/854772/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/854772/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/854772/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/854772/3"}}},"commit":{"parents":[{"commit":"623fd75d394e802011aa5908be9d3aa236dbe46f","subject":"Merge \"Fix a deprecation warning about threading.Thread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/623fd75d394e802011aa5908be9d3aa236dbe46f"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-08-26 11:16:06.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-08-26 15:56:57.000000000","tz":60},"subject":"Remove \u0027glance-download\u0027 from default import methods","message":"Remove \u0027glance-download\u0027 from default import methods\n\nAs \u0027glance-download\u0027 requires federated deployments it should\nnot be enabled by default.\n\nAdded line explaining it being available for such deployments\nin the config help text.\n\nChange-Id: Icc792e23aad9e7a9a788e6b5826bc5aae54ae978\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c342c0e944126c6d257fab8910477a58bdec5da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c342c0e944126c6d257fab8910477a58bdec5da7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}}]}],"submit_requirements":[]}
