)]}'
{"id":"openstack%2Fglance~854938","triplet_id":"openstack%2Fglance~master~Ie229c57654a8827d2274f6ab812310bdd989db4b","project":"openstack/glance","branch":"master","topic":"jobs","attention_set":{},"removed_from_attention_set":{"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2022-12-15 15:50:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie229c57654a8827d2274f6ab812310bdd989db4b","subject":"add openstack-python3-zed-jobs-arm64 job","status":"MERGED","created":"2022-08-29 03:45:29.000000000","updated":"2022-12-15 15:52:46.000000000","submitted":"2022-12-15 15:50:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854938-jobs","meta_rev_id":"a7163a7bb11b83e842dfc90278ca7d164680346d","_number":854938,"virtual_id_number":854938,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-15 15:50:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-10-20 21:04:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-10-26 11:35:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-12-15 14:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-21 03:14:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-12-15 14:33:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-10-20 21:04:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 03:57:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-29 04:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 21:04:57.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-10-21 03:14:42.000000000","updated_by":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"reviewer":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"state":"REVIEWER"},{"updated":"2022-10-26 11:35:32.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-12-15 14:33:46.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"c25baae14e92074fa0f2f44c941420493952e558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-08-29 03:45:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3a3bdf5393618da32bff532a1f542cbdf51aa03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 03:57:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/827394c26f8646d8bcd9ec4d87455531\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/787b4f3c37904a4a98632af32734cf65 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8e1851f2db04bafab2be14f33e48afd : SUCCESS in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ab8733395d6d711e168357cd42f9fe1ee3b2baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 04:28:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/411dfa1e0f704bf087f796198c26b381\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63c617d23371443a981b5301582a6b32 : SUCCESS in 41m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba68ef62de24fd8a1d4d697ec864165 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97e78e61805643479e80a396e03ed252 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/777868a0e74145b58528b4d455b199ea : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c91825efe7e5414fbde4c08d82c4e35d : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a635a0b0cfd84a33aff35058e06c1257 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/76e2f8400c544e1c975c8737481382fc : SUCCESS in 11m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/154e3200f69146c6b2d8dcdcc0d12fcc : SUCCESS in 8m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/286f6ea2aaf9435e84ae20a55dc90769 : SUCCESS in 8m 07s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dab36b46525d444baee3bfcd3e0f01ba : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"29697718f07dea6f188125d4f9682109c232a7fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-10-10 01:38:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3eec4e03117de6935e069b8c99cb2ad0f1554590","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 01:50:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b59d0621ef740058a62203e9338843d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/817b82ea807740878df9910b53644256 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2ba67162d8ac42be84af9fc0e47f009c : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2609825bf410e111c0c9a8eec17f864162968093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 02:31:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7da02099130b48d1863fa55a5741a623\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2ae39c4005e74dd0894a5f650724fa81 : SUCCESS in 51m 09s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27bb73483eb942f2968acc49106aef4f : SUCCESS in 46m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5f63cbfd3a460fbbcb95c81d9069bb : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc51a2bf434b4fef92b5383c9395d8d4 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11c22240481a4fdf9b91358d52f69bf6 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83989143a5bb41c1b74a4325c3d6623e : SUCCESS in 6m 49s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3601282a8378445296474511c218ebcb : SUCCESS in 10m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/656bbff0ff1347d283c6b7eba39fa832 : SUCCESS in 8m 14s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e10ec4c68ecf4d62b75b71ef9481722e : SUCCESS in 9m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3cae08d27a314566af21226c092a69b2 : SUCCESS in 23m 04s","accounts_in_message":[],"_revision_number":2},{"id":"979884ff2eea16b538bc14df52375f531f4f23a1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-10-20 21:04:57.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a5a72bb300c6e69e26f19f92a112cf3aa63e3ed","author":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"date":"2022-10-21 03:14:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5395d30519941fb8b2d236aa759d32788051c3d1","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-10-26 11:35:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"874de17eb08fa913122736c837986ec5f2d2d1c7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-12-15 14:33:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a555354e2c84e22498a42702b5cebdb8193a8273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 14:34:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3f405d041c0878e2fbf41cfd13f6932286421da6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 15:50:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cad5c6f532e042d5bd6bd68d51938eb7\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d69775eaa2c428e88c606b7ce9a6711 : SUCCESS in 1h 09m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614d5bf1f24f4ad99d3238acacecf569 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdd7f0dab8284703be642c1fd550300e : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b529d386b796463492a8f215fd5a64cd : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c10085507584fb0a02c3ae4632a7316 : SUCCESS in 6m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c3ba48f6fafb42998846020bb56d7b40 : SUCCESS in 8m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb88d86bef4a4818a33d96344cf90d0c : SUCCESS in 8m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9caf4a94498c4d05b620765657700f53 : SUCCESS in 22m 05s","accounts_in_message":[],"_revision_number":2},{"id":"d7f97b04183f4ce50ac29171565cd033f7d2ad79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 15:50:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a7163a7bb11b83e842dfc90278ca7d164680346d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 15:52:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89102d03d0a64e5c9c30912d290c248b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f4643c8e10749a7b0b81b50977e3099 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4da0077d7bec8eff31dd673e8a6d1b5bcabffb89","revisions":{"30fbcc5a94b10401ebaabf67f298bc285bc53f84":{"kind":"REWORK","_number":1,"created":"2022-08-29 03:45:29.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/38/854938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/854938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/854938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/854938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/854938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/854938/1"}}},"commit":{"parents":[{"commit":"c342c0e944126c6d257fab8910477a58bdec5da7","subject":"Remove \u0027glance-download\u0027 from default import methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c342c0e944126c6d257fab8910477a58bdec5da7"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2022-08-29 03:45:09.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2022-08-29 03:45:09.000000000","tz":480},"subject":"add openstack-python3-zed-jobs-arm64 job","message":"add openstack-python3-zed-jobs-arm64 job\n\nThis is a non-voting job to validate py3 unittests on ARM64\n\nChange-Id: Ie229c57654a8827d2274f6ab812310bdd989db4b\nTask: #40402\nStory: 2007938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30fbcc5a94b10401ebaabf67f298bc285bc53f84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30fbcc5a94b10401ebaabf67f298bc285bc53f84"}]},"branch":"refs/heads/master"},"4da0077d7bec8eff31dd673e8a6d1b5bcabffb89":{"kind":"REWORK","_number":2,"created":"2022-10-10 01:38:34.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/38/854938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/854938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/854938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/854938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/854938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/854938/2"}}},"commit":{"parents":[{"commit":"dc11640d2fea4010cd64f3bac861ab8bc3501044","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc11640d2fea4010cd64f3bac861ab8bc3501044"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2022-08-29 03:45:09.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2022-10-10 01:38:19.000000000","tz":480},"subject":"add openstack-python3-zed-jobs-arm64 job","message":"add openstack-python3-zed-jobs-arm64 job\n\nThis is a non-voting job to validate py3 unittests on ARM64\n\nChange-Id: Ie229c57654a8827d2274f6ab812310bdd989db4b\nTask: #40402\nStory: 2007938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4da0077d7bec8eff31dd673e8a6d1b5bcabffb89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4da0077d7bec8eff31dd673e8a6d1b5bcabffb89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[]}
