)]}'
{"id":"openstack%2Fglance~859350","triplet_id":"openstack%2Fglance~stable%2Fyoga~I75552035694a56601ad5256a3ba3c520702c16a3","project":"openstack/glance","branch":"stable/yoga","topic":"dnm-yoga","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-09-28 02:42:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I75552035694a56601ad5256a3ba3c520702c16a3","subject":"DNM: Test whether py36 still passes on this branch","status":"ABANDONED","created":"2022-09-26 22:48:15.000000000","updated":"2022-09-28 02:42:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"598f67b3144a746f5ddf230673942e4364e90c7e","_number":859350,"virtual_id_number":859350,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-27 00:30:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-27 00:30:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0617bb48f9a99481ae06c15ae1ca6be2b9184603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-09-26 22:48:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e3d81aa8a798ccdd7b9a5efd0613681217ef90e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 00:30:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/131897364e96447aabc3b00fe2c13e77\n\n- grenade https://zuul.opendev.org/t/openstack/build/b0ea3265a44d41cc9c5232c2a59c15d3 : SUCCESS in 41m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b7827f069dd24e4c8ae78b9f055a7995 : SUCCESS in 47m 42s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48404f592e01421c9b285d3f13fb36fd : SUCCESS in 1h 20m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac892c7f523a4f4baec68f339f8edd36 : SUCCESS in 51m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd1f876c14c4982804859018d7aabfd : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df4d5be4b1f7472eb7678d331b1b6ec7 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/026a93afdcad48129622aac2de6ca9e9 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a582a03fc5404e7887c9aa7a9ebbb997 : SUCCESS in 5m 41s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/525aabd6433e4e5284d42462b51c71d5 : SUCCESS in 11m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f5f5e085d25c44c296c0380490374f3e : SUCCESS in 9m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a13a82e50fa3444b9469ea3c4d38f451 : SUCCESS in 9m 47s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6a3e4790101b4becaddd030ac772c427 : SUCCESS in 5m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dafd7ebdc2574d46a0f0ac5269bea22f : SUCCESS in 1h 12m 03s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/93c1b577d7834569b4e82a0f1ebe270c : SUCCESS in 1h 25m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b6386237aa9d4136986080d8619573be : SUCCESS in 1h 28m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1459b4f9b0e84d9e97355eaf1916f454 : SUCCESS in 1h 40m 19s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/4540330a1deb45e1a6305775b1cd5dd7 : RETRY_LIMIT in 21m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9a1d863ab5b4678b73ced8745d4ed33 : SUCCESS in 50m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/63c717740dbc42f989d408c0827c4048 : SUCCESS in 1h 08m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cf1056a581ca4599be5e9db82ea276d6 : FAILURE in 33m 32s","accounts_in_message":[],"_revision_number":1},{"id":"598f67b3144a746f5ddf230673942e4364e90c7e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-09-28 02:42:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c5fb0b34f5f37120bbc9e1d7cee5196e66b5cc7","revisions":{"5c5fb0b34f5f37120bbc9e1d7cee5196e66b5cc7":{"kind":"REWORK","_number":1,"created":"2022-09-26 22:48:15.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/50/859350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/859350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/859350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/859350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/859350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/859350/1"}}},"commit":{"parents":[{"commit":"f3ae67fa1925375fef6af44e73db201db4fe359c","subject":"Mock getaddrinfo in cinder store tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3ae67fa1925375fef6af44e73db201db4fe359c"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-09-26 22:47:54.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-09-26 22:47:54.000000000","tz":120},"subject":"DNM: Test whether py36 still passes on this branch","message":"DNM: Test whether py36 still passes on this branch\n\nChange-Id: I75552035694a56601ad5256a3ba3c520702c16a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c5fb0b34f5f37120bbc9e1d7cee5196e66b5cc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c5fb0b34f5f37120bbc9e1d7cee5196e66b5cc7"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
