)]}'
{"id":"openstack%2Fglance~861374","triplet_id":"openstack%2Fglance~master~I226ca5b77345ff1284d3cc09b4a719f1373ea084","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"34325":{"account":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"last_update":"2022-11-03 09:08:08.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-11-02 13:30:19.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}}},"hashtags":[],"change_id":"I226ca5b77345ff1284d3cc09b4a719f1373ea084","subject":"Fix a document description error","status":"MERGED","created":"2022-10-14 13:17:16.000000000","updated":"2022-11-03 09:09:15.000000000","submitted":"2022-11-03 09:08:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861374","meta_rev_id":"6d751a744867c60ee1cfd343e77f2531f1726986","_number":861374,"virtual_id_number":861374,"owner":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-03 09:08:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"all":[{"value":1,"date":"2022-11-03 06:57:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2022-11-02 07:47:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},{"value":2,"date":"2022-11-03 08:54:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-11-02 13:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},{"value":1,"date":"2022-11-03 08:54:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-14 13:29:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-02 07:47:00.000000000","updated_by":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-11-03 06:57:25.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2022-11-03 08:54:16.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"0de448fca431223752a65f119d24dddff4aa7414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"date":"2022-10-14 13:17:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1e3fa451b81d0bd6a52c4de5c52ffb0d2f57a61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"date":"2022-10-14 13:18:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a3d2337883895328febc84dc67a10b6dedc811f","author":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"date":"2022-10-14 13:23:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9907fe4ad59c748b8a808d086251615b98b14f20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 13:29:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04e3a4cdf1bf4e9f937533f2558f1549\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32ab177aa6ca4264bec4b84ab4a54bb4 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f54e242e2ec6467ebce00c8d003c8a16 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":2},{"id":"dc04c5d846c9366d16c94d263ea3a193f498716a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"date":"2022-10-14 13:37:08.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nEdit commit msg","accounts_in_message":[],"_revision_number":3},{"id":"1a118edcfac7ea77269d34ea9cffc07bdf7988fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 13:47:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b942459d40d047f9aa7fa4ba61face8f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d875bb30dcde445abd11353869d30eac : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/411c87967564456583f9283222d227ce : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":3},{"id":"5710e0137142997d615643475bda6167709a95ff","author":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"date":"2022-11-02 07:47:00.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4ab538fcfcf4a20e4d7fca9f369fc79d198a4120","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-02 13:30:19.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82a12ec3e9fce1a8671c2ee6e52d6b826dabbd82","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-11-03 06:57:25.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4ebadc3b1bbcac019cc5dedf77eae794602978d8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-11-03 08:54:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4ed371ed6b1e6582e919b66a0a544d59568e56de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 08:54:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6fce9c0fc83195585811c94b5f6da3f0081842f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 09:08:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95e5228b8b4f4d54aaa0fdb83618de87\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d36873374d024a6cb4a9be0a490f48d6 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8af6ef75cbe148b58afa8e60c6a62b20 : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":3},{"id":"11aafa5dc2d9f76dae1ff01211344eea912d60fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 09:08:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6d751a744867c60ee1cfd343e77f2531f1726986","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 09:09:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e93d885b52646ea8ca7a04df633d0e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae0cf16712a4d5a93db9a87b2fa1247 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bbc0df8d40f0b2ae2887e8133135997757aeaaa1","revisions":{"564211810544196a7c2c525ce95f20b49180218a":{"kind":"REWORK","_number":1,"created":"2022-10-14 13:17:16.000000000","uploader":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"ref":"refs/changes/74/861374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/861374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/861374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/861374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/861374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/861374/1"}}},"commit":{"parents":[{"commit":"3224e05ef4cee5f74b0a996aaac4bb6be13b114b","subject":"Merge \"Fix a wrong assertion method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3224e05ef4cee5f74b0a996aaac4bb6be13b114b"}]}],"author":{"name":"jneeee","email":"jneeee@qq.com","date":"2022-10-14 12:59:18.000000000","tz":480},"committer":{"name":"jneeee","email":"jneeee@qq.com","date":"2022-10-14 13:16:25.000000000","tz":480},"subject":"Change a doc description error","message":"Change a doc description error\n\nuse_keystone_quotas should be use_keystone_limits. Just like in\nElsewhere.\n\nCloses-bug: #1992974\nChange-Id: I226ca5b77345ff1284d3cc09b4a719f1373ea084\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/564211810544196a7c2c525ce95f20b49180218a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/564211810544196a7c2c525ce95f20b49180218a"}]},"branch":"refs/heads/master"},"cac8a7269a893141263f75569b0727e21b8ab141":{"kind":"REWORK","_number":2,"created":"2022-10-14 13:18:54.000000000","uploader":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"ref":"refs/changes/74/861374/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/861374/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/861374/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/861374/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/861374/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/861374/2"}}},"commit":{"parents":[{"commit":"3224e05ef4cee5f74b0a996aaac4bb6be13b114b","subject":"Merge \"Fix a wrong assertion method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3224e05ef4cee5f74b0a996aaac4bb6be13b114b"}]}],"author":{"name":"jneeee","email":"jneeee@qq.com","date":"2022-10-14 12:59:18.000000000","tz":480},"committer":{"name":"jneeee","email":"jneeee@qq.com","date":"2022-10-14 13:18:39.000000000","tz":480},"subject":"Change a doc description error","message":"Change a doc description error\n\nuse_keystone_quotas should be use_keystone_limits. Just like in\nElsewhere.\n\nCloses-bug: #1992974\nChange-Id: I226ca5b77345ff1284d3cc09b4a719f1373ea084\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cac8a7269a893141263f75569b0727e21b8ab141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cac8a7269a893141263f75569b0727e21b8ab141"}]},"branch":"refs/heads/master"},"bbc0df8d40f0b2ae2887e8133135997757aeaaa1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-10-14 13:37:08.000000000","uploader":{"_account_id":34325,"name":"Jneeee","email":"jneeee@qq.com","username":"junnan.liu"},"ref":"refs/changes/74/861374/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/861374/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/861374/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/861374/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/861374/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/861374/3"}}},"commit":{"parents":[{"commit":"3224e05ef4cee5f74b0a996aaac4bb6be13b114b","subject":"Merge \"Fix a wrong assertion method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3224e05ef4cee5f74b0a996aaac4bb6be13b114b"}]}],"author":{"name":"jneeee","email":"jneeee@qq.com","date":"2022-10-14 12:59:18.000000000","tz":480},"committer":{"name":"jneeee","email":"jneeee@qq.com","date":"2022-10-14 13:35:59.000000000","tz":480},"subject":"Fix a document description error","message":"Fix a document description error\n\nuse_keystone_quotas should be use_keystone_limits. Just like in\nElsewhere.\n\nCloses-bug: #1992974\nChange-Id: I226ca5b77345ff1284d3cc09b4a719f1373ea084\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbc0df8d40f0b2ae2887e8133135997757aeaaa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbc0df8d40f0b2ae2887e8133135997757aeaaa1"}]},"branch":"refs/heads/master","description":"Edit commit msg"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
