)]}'
{"id":"openstack%2Fglance~861752","triplet_id":"openstack%2Fglance~master~Ic97825dd9c38ec1c759e5ca610a19c93fe4ca6a6","project":"openstack/glance","branch":"master","topic":"db_constant_fix","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-02-08 13:49:52.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-02-08 15:45:16.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-02-02 20:30:29.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ic97825dd9c38ec1c759e5ca610a19c93fe4ca6a6","subject":"Update migration constant","status":"MERGED","created":"2022-10-18 13:01:48.000000000","updated":"2023-02-08 15:46:54.000000000","submitted":"2023-02-08 15:45:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861752-db_constant_fix","meta_rev_id":"f9acc3b83e3903ec25037dd8471aeed2cda4996d","_number":861752,"virtual_id_number":861752,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-08 15:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2023-02-06 13:40:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2023-02-08 13:49:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-02-02 20:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2023-02-08 13:49:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2023-01-18 05:16:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 14:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-18 05:16:53.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-01-18 05:16:53.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-02-06 13:40:17.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"d6ad0f1a062953af1c27fdc5fc51c27e25cc1879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-10-18 13:01:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04e49796978072c946beace358e4401ee62388a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-10-18 13:13:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"870133602b8cae67b0d698218fc01985e2e93cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 14:49:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d72803a784a4a479a5112973b6a27ec\n\n- grenade https://zuul.opendev.org/t/openstack/build/8c5fd436c0cc4d74b1a387516abbf18b : SUCCESS in 49m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/385ced80870e4d1c83bf4c6b47016c13 : SUCCESS in 50m 02s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d58cf83e5cdd4b20b0085ee6ac54611b : SUCCESS in 1h 26m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f335be8d84044f782570c64df9061e4 : SUCCESS in 50m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54ddbb6d33dc4dc0b0c57b2c53acb014 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/084ee49c9dfb4e0791977f312cb9efaf : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8aecdb5416aa4690847d54cae24373d7 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29367511ebde4477b01d83dd9f86226c : SUCCESS in 6m 46s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/16221461906949a7934777b091674cb1 : FAILURE in 12m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7cfac83ea0ed478583a877282bc917f5 : FAILURE in 10m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3b73e3c4df3d410e8f2bcaf122f9682c : TIMED_OUT in 31m 08s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/73cc36cbdd4d4ccfb8c652e735c68ea2 : FAILURE in 4m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4217f51ffad64dcab0dec93929cd753c : SUCCESS in 1h 17m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c9a0e8dc606f40aa880bb3a0ac9d71fd : FAILURE in 1h 12m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b1bbd28b625d49f5ab55361d6e98149e : SUCCESS in 1h 12m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f12379a4ceed49faa95e8fb42ff018c1 : SUCCESS in 1h 28m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8069d74838d40c6a6af3ec0341351cb : SUCCESS in 41m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/be6469287af54b189a9ff0d674a57500 : FAILURE in 1h 16m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a507d5d9fe114ca6be9bebfe496a6867 : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":2},{"id":"5a4a726c7661331de21a008c783d8deb2d49c200","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-10-21 09:20:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"448542c34cafe6577723a123e9a073ee9aa5241e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-11-03 13:56:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9470893ac3aa2e161e7718af8e7bbf5c4e9733b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 15:42:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/540ab318f8f04cf882dc0264e4b5f026\n\n- grenade https://zuul.opendev.org/t/openstack/build/abb50133c91d485cbc2b1cfd2fcf367a : SUCCESS in 42m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/354486635d5e473e8fc37cfa8a227e91 : SUCCESS in 49m 49s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fe758b4d3bed42e4a32c5483e2a39c50 : SUCCESS in 1h 15m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a60d4545bf544498a97a2d44c5cc020 : SUCCESS in 43m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80bf5a90b1a04316879c2e46351143a8 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7677c0fc4b14431eb49ac6b99bca916a : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/838a435a540d4a07acf7053b1d7a4c7b : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f03d26d16a44898b16ac294ee0d0d20 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/478ac01970514099a23104e338529a54 : FAILURE in 10m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8311d8aea8174301ada47047b7272688 : FAILURE in 7m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c6f7fe7c8b4149b386ce6122a62d4d12 : FAILURE in 8m 01s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d5052f4a0032477caca72bb0d5be824f : FAILURE in 5m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/725f8ee3382f4e83a11a3601de8e238a : SUCCESS in 1h 02m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2eda08fb173c4af084b2b8f5233f36f8 : SUCCESS in 1h 10m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/97854c3c256a4992b5ade2d7178044d7 : SUCCESS in 1h 08m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4e44db72b1a241bf8901aa8b4ce47100 : SUCCESS in 1h 40m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e52b767b42e4ae3a36b35f7cdf59062 : SUCCESS in 43m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b2ab921f646c4ba0944fe9309bdf5d2c : SUCCESS in 58m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/243b7bbb681146189b72a2db69f9c88f : SUCCESS in 26m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a8d7cf01ad0fd6e6113e03d64bce5d58f60e8087","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-11-08 08:47:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f0bb2a0bfd912b709130f82bc32c2843132e8ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 10:31:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa62706b94ec424cb859eb91c385dd8a\n\n- grenade https://zuul.opendev.org/t/openstack/build/16d0e74ead9b469fa8043e7d7458dd86 : SUCCESS in 49m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/839e0ed196254cbe85b4955c548dc0d9 : SUCCESS in 52m 26s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab0c08d2221643a89aae9ca5d2235dd2 : SUCCESS in 1h 28m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87be813ea1fb424f85ca0b6519b2e4ed : SUCCESS in 47m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/600ffc2171f14883a05d5227b4002dba : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/520b15818e7a4f7d88e9c24f949e42db : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/344917c3ad6e460fa1bd036026340f78 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4de321aa1f344a6bbae8191774e4d8b : SUCCESS in 5m 54s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/619eef6ec77c481293a0302f7263e4da : FAILURE in 11m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea311bcc830f439797b04803643feb1d : FAILURE in 10m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4dbde95a33ab46f4a1babcfba48723e3 : FAILURE in 8m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1f851e4217e74e84970a2a6458349037 : SUCCESS in 5m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dbd4b0fe11ca4f449a535b2f2a63eb83 : FAILURE in 1h 15m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/be6f6ad0dfab4de1a2292d87010b7b22 : SUCCESS in 1h 32m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/146ee72c96694c1d8e13315f86575739 : SUCCESS in 1h 31m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e98b1b4e1a1745ce8b5f4d0f90d7ebd6 : SUCCESS in 1h 36m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdd76d46098f4269a69f295521428e96 : SUCCESS in 58m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0976a8f53d6a495bb7d3b16097d2606f : SUCCESS in 54m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f2f9f49bb7d847e788bb55a0151faa87 : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":3},{"id":"d409101735bc26d440d93580a0e6a63848b0ace8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-11-08 12:12:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a99bbcea121b50371493bfe93c0e3869e3087a4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 13:45:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61380a2170774fbdbee05ef180185e19\n\n- grenade https://zuul.opendev.org/t/openstack/build/570a2c97542049ca8565e4ce2df5e239 : SUCCESS in 51m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7df03a7dfcf54785afac9e506f3e3d8c : SUCCESS in 49m 17s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ded43d53d214f9a9b7949fd3bdd26a5 : SUCCESS in 1h 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/237f8d16c9c54425ba409a4f5de5a731 : SUCCESS in 47m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e6c54feb3954acb895663da3ebc22bc : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4dcf172c4eb4dc4bcc0a9269cd0d308 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5aadb4b0bf4b4eb48d35e5f3941185f9 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/798a19551764469e9461d9064ad98279 : SUCCESS in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/16e97ec6fae14424b6678862c2dcce33 : SUCCESS in 11m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/91420689d1c3491e9947091899eca0a4 : SUCCESS in 10m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/23e23724bfba4837b481dee5f6e593ba : SUCCESS in 9m 34s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f311ab969df0422aa08d3a0cfd46e760 : SUCCESS in 4m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/41d0d557ee0847d5a968a43437336585 : SUCCESS in 1h 06m 30s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4aa79a4a26df4816a8660146afee96c4 : SUCCESS in 1h 08m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/94b075b69a8a4e4183d4d65e96c48682 : SUCCESS in 1h 17m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a065553b04844ac3811762e0e420cf83 : SUCCESS in 1h 31m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b49f0a759d194659b8a8f9b3faa1b32d : SUCCESS in 56m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bfb0f0cdc48945a198d0de2ad7fe6051 : SUCCESS in 1h 04m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5c98c8dfc81041ec9153a0dfa6655deb : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":4},{"id":"80ca2b888337082196496f6ae10b23baf170688e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-01-16 14:58:11.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"ffba5c2e98bd290177a1dcc6e92b0916d6dd0c6c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-01-16 14:58:36.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"916918b49d8c6a6d6f21f09336c88948419596ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 14:58:39.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 861752,5","accounts_in_message":[],"_revision_number":5},{"id":"770b03ff2730e8bd4686fdd1ccb47a19bc9d5f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 14:59:31.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 861752,5","accounts_in_message":[],"_revision_number":5},{"id":"c356256cf0df0dfadd1308a88c7ae5642ae80943","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-01-18 05:16:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8398f952e8f1bf53c7884bb98a361d319c657ee7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-01-18 05:16:53.000000000","message":"Patch Set 6: Review-Priority+2","accounts_in_message":[],"_revision_number":6},{"id":"60f28888810f5ddc065f505d1bc9c01adc4a096f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 06:22:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/728a6f590abd44a2874d02105cdd1a89\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/468f567ded1d44f7ae654ba595503a6b : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d6d7b804780b4c52942a82bfcb857564 : SUCCESS in 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"906382527841d5ca26fba18229b3430b2c5449f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 07:00:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02392efcf99f4200a81131c64df6e6e4\n\n- grenade https://zuul.opendev.org/t/openstack/build/b727a9634155431a9f577a846c9fdc3f : SUCCESS in 1h 01m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ef3fe0ee296247fb87aeeac6e41090c2 : SUCCESS in 1h 05m 16s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d5d41b4054842d88a564a34491f5dc4 : SUCCESS in 1h 30m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac94337df8914c30b2e9e2e4229ed9b0 : SUCCESS in 1h 00m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a4e835b743a49abb429fa7849de56d3 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3588828dc35d465e879deccafb9fbeae : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40dde35c6cd84306805ffef4c7cf7759 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7572709238b8483ea32155ab937e9714 : SUCCESS in 6m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/adca5d9ac0564634a4383c60a1392939 : SUCCESS in 12m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9454d42906ea4b72a06ee2e5da9dc4c7 : SUCCESS in 7m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/51c10a0f1e9c441e8f681a63a9ed5c09 : SUCCESS in 8m 03s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3b9123388a0549d98789444dcdaf6887 : SUCCESS in 1h 13m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c087128c697140c4adf2f772cb6afcce : SUCCESS in 1h 24m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/21cff5bcb02b4758abae64ee6d8ea323 : SUCCESS in 1h 37m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2dbd33285fbc46659e47573d8c18ad16 : SUCCESS in 1h 37m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5370d03201cb4b5895a15f3206878133 : SUCCESS in 48m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6b74dea2465c40b58ef5d8af048bd438 : SUCCESS in 51m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f7ab9ac8d5524217a5ce118d5d2a6f16 : SUCCESS in 36m 52s","accounts_in_message":[],"_revision_number":6},{"id":"7c038d2081f6e7a2702fcf31d2261ffbf14fcab3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-02 20:30:29.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"55c2570e707c294515f28752e9aa93b64efcbc06","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-06 13:40:17.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"6764c39df48a6ba49582e50be963f2dc93fb1cad","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-02-08 13:49:52.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"48bbd8a3e692debd730701acaac5ed71fb10c519","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 13:50:33.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"87c4b3b999a5672196fbb14d7066c365aae38eca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 15:45:16.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/708ad8236153463bbf94e1b5b51625a6\n\n- grenade https://zuul.opendev.org/t/openstack/build/6c162a9ba0e04901b62d41cc6586abcc : SUCCESS in 1h 02m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a5eb2c17a8644e4a30a63168317ba7b : SUCCESS in 1h 40m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f44aac0666ed42b2ae750be05ba83dcb : SUCCESS in 1h 06m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aba4c3a28f84572a61a9229ce6129f8 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b8600d91fdf4071bd98bd508330569b : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dda9d7f3d3e84f4880ef45297a5ad752 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13a54e05089453eb33290bd7480dbb1 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/70b8b5192477431887377da514ca4373 : SUCCESS in 7m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a642464ef4744bebf6f2b90ad1d186b : SUCCESS in 9m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cc00935b642c45e2acbdd3425e137031 : SUCCESS in 1h 48m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/beae0247029b4b9eb40cafbf82eddcc5 : SUCCESS in 1h 25m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a38b3e0bb94a42f6aee06341c3be9e2d : SUCCESS in 1h 00m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6a5c17ee13f4df2ae40b07c3672e5ca : SUCCESS in 51m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a4045208e20747459a4c4dc5d3484e08 : SUCCESS in 1h 22m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a4373eedd9a74b7da10c5c25597025f2 : SUCCESS in 39m 09s","accounts_in_message":[],"_revision_number":6},{"id":"0d27783f33178f8e9e695bed1635e65638ed2506","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 15:45:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"f9acc3b83e3903ec25037dd8471aeed2cda4996d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 15:46:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63029a6ef536445c9d5dfe9396036cf7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b62221c29154f59b7f45b94b5f8132f : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9786936915f7745c2b9109eb475962f41a0df67b","revisions":{"6adbb3e139c6a6cbdabd46b89fa022223077ca22":{"kind":"REWORK","_number":1,"created":"2022-10-18 13:01:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/861752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/861752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/861752/1"}}},"commit":{"parents":[{"commit":"3224e05ef4cee5f74b0a996aaac4bb6be13b114b","subject":"Merge \"Fix a wrong assertion method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3224e05ef4cee5f74b0a996aaac4bb6be13b114b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-10-18 12:51:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-10-18 12:56:12.000000000","tz":0},"subject":"[WIP] Update migration constant","message":"[WIP] Update migration constant\n\nUpdate the data migration current release to \u00272023.1\u0027.  Include a\nsemver pseudo-header in this commit message so that pbr will\nincrement the major version number, otherwise\nglance.tests.unit.gate.test_data_migration_version will break.\n\nSem-Ver: api-break\nChange-Id: Ic97825dd9c38ec1c759e5ca610a19c93fe4ca6a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6adbb3e139c6a6cbdabd46b89fa022223077ca22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6adbb3e139c6a6cbdabd46b89fa022223077ca22"}]},"branch":"refs/heads/master"},"84f0c057939753f67122bfffb088a387d8de67d7":{"kind":"REWORK","_number":2,"created":"2022-10-18 13:13:53.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/861752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/861752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/861752/2"}}},"commit":{"parents":[{"commit":"3224e05ef4cee5f74b0a996aaac4bb6be13b114b","subject":"Merge \"Fix a wrong assertion method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3224e05ef4cee5f74b0a996aaac4bb6be13b114b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-10-18 12:51:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-10-18 13:08:05.000000000","tz":0},"subject":"[WIP] Update migration constant","message":"[WIP] Update migration constant\n\nUpdate the data migration current release to \u00272023.1\u0027.  Include a\nsemver pseudo-header in this commit message so that pbr will\nincrement the major version number, otherwise\nglance.tests.unit.gate.test_data_migration_version will break.\n\nSem-Ver: api-break\nChange-Id: Ic97825dd9c38ec1c759e5ca610a19c93fe4ca6a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/84f0c057939753f67122bfffb088a387d8de67d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/84f0c057939753f67122bfffb088a387d8de67d7"}]},"branch":"refs/heads/master"},"feb602fd906add885b61704befe7848af9308b58":{"kind":"REWORK","_number":3,"created":"2022-11-08 08:47:44.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/861752/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/861752/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/861752/3"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-10-18 12:51:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-11-08 08:40:58.000000000","tz":0},"subject":"[WIP] Update migration constant","message":"[WIP] Update migration constant\n\nUpdate the data migration current release to \u00272023.1\u0027.  Include a\nsemver pseudo-header in this commit message so that pbr will\nincrement the major version number, otherwise\nglance.tests.unit.gate.test_data_migration_version will break.\n\nChange-Id: Ic97825dd9c38ec1c759e5ca610a19c93fe4ca6a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/feb602fd906add885b61704befe7848af9308b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/feb602fd906add885b61704befe7848af9308b58"}]},"branch":"refs/heads/master"},"ba1f3eca404ac368e0a73e1b715bc4a92ed6f37b":{"kind":"REWORK","_number":4,"created":"2022-11-08 12:12:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/861752/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/861752/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/861752/4"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-10-18 12:51:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-11-08 12:05:42.000000000","tz":0},"subject":"[WIP] Update migration constant","message":"[WIP] Update migration constant\n\nUpdate the data migration current release to \u00272023.1\u0027.  Include a\nsemver pseudo-header in this commit message so that pbr will\nincrement the major version number, otherwise\nglance.tests.unit.gate.test_data_migration_version will break.\n\nChange-Id: Ic97825dd9c38ec1c759e5ca610a19c93fe4ca6a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba1f3eca404ac368e0a73e1b715bc4a92ed6f37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba1f3eca404ac368e0a73e1b715bc4a92ed6f37b"}]},"branch":"refs/heads/master"},"24b9e96dd5782f2aac2747e572b27657c9eb8e0c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-01-16 14:58:11.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/861752/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/861752/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/861752/5"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-10-18 12:51:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-01-16 14:58:11.000000000","tz":0},"subject":"Update migration constant","message":"Update migration constant\n\nUpdate the data migration current release to \u00272023.1\u0027.\nDepends-On: https://review.opendev.org/c/openstack/glance/+/870549\n\nChange-Id: Ic97825dd9c38ec1c759e5ca610a19c93fe4ca6a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/24b9e96dd5782f2aac2747e572b27657c9eb8e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/24b9e96dd5782f2aac2747e572b27657c9eb8e0c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9786936915f7745c2b9109eb475962f41a0df67b":{"kind":"REWORK","_number":6,"created":"2023-01-18 05:16:09.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/861752/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/861752/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/861752/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/861752/6"}}},"commit":{"parents":[{"commit":"d07187f7107b02f42cfa335a22763cbb1c085587","subject":"Remove migration constant job and test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d07187f7107b02f42cfa335a22763cbb1c085587"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-10-18 12:51:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-01-18 04:57:34.000000000","tz":0},"subject":"Update migration constant","message":"Update migration constant\n\nUpdate the data migration current release to \u00272023_1\u0027. Added\nempty migration scripts to rule out regression scenarios as well.\n\nChange-Id: Ic97825dd9c38ec1c759e5ca610a19c93fe4ca6a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9786936915f7745c2b9109eb475962f41a0df67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9786936915f7745c2b9109eb475962f41a0df67b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[]}
