)]}'
{"id":"openstack%2Fglance~862285","triplet_id":"openstack%2Fglance~master~I500f85b42873239d751031d1714eff989e463f93","project":"openstack/glance","branch":"master","topic":"inject-properties-upload","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2022-10-23 20:34:27.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I500f85b42873239d751031d1714eff989e463f93","subject":"WIP: Use inject metadata properties in upload","status":"ABANDONED","created":"2022-10-21 09:53:00.000000000","updated":"2023-04-11 21:12:45.000000000","total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"37d95e1d8cef89de9ce302fc7d1fb0e8dcf13893","_number":862285,"virtual_id_number":862285,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-21 11:19:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 11:19:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e504a6bb6627e6bdf03ed1512ae07fae65378599","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-21 09:53:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50540b73ec43e7b336d8440fa4b44d4a7ae8f018","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-21 09:59:53.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f2b956483dac29e61c0ed4d1cb5b065f45028bd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 11:19:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/306fac5182ff4d6dbad6b0be62574fd4\n\n- grenade https://zuul.opendev.org/t/openstack/build/6133d2345d174830b5a94e37ec0d63aa : FAILURE in 47m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2b336a82dfef466f962a559fe0be77b6 : FAILURE in 49m 02s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb97c8958d0d40369c41a30e966077b3 : FAILURE in 1h 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73acee7e24634d20a6a862667b09f777 : POST_FAILURE in 23m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e7dd11c55646ed8f53080bd30d03fa : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e99e6b87eb3e4464a17c7513a66f0e63 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/868e550781a3469c94ad362ca25335ad : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8102839deb8e4de2a9933244b3f6fd44 : SUCCESS in 6m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cf03d13b43ee40cb9e532115c92e3d8f : FAILURE in 10m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6038553f791441f1bd94d4ace967e48d : FAILURE in 7m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/412161dc6a7b4cc89b49f5ae44571d80 : FAILURE in 7m 24s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d9277b5c5dd24754802a31321710e030 : SUCCESS in 4m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d34d33c0c1b943eaae8267248dd8cf3e : FAILURE in 1h 18m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b87d5e4f52724a29b60bcee3006787c6 : FAILURE in 1h 03m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a4f16993586047b6a90cfbbdf0f1226a : FAILURE in 1h 20m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e68b9a9f02c24f0abc866917074416f6 : FAILURE in 1h 15m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a53a9e058cbc4601977c3f9ba5c4c0a2 : FAILURE in 21m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/09e48b90880c4f8eb343590ccf954b70 : FAILURE in 1h 14m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4b565f16f52b4032bb2810f29d9840a7 : FAILURE in 22m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f182e016e705079683891d82ce76399843871e21","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-23 20:34:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37d95e1d8cef89de9ce302fc7d1fb0e8dcf13893","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-11 21:12:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8177f388c79f30970b16e315ddfeb133f98005f7","revisions":{"8177f388c79f30970b16e315ddfeb133f98005f7":{"kind":"REWORK","_number":1,"created":"2022-10-21 09:53:00.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/85/862285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/85/862285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/862285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/862285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/862285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/85/862285/1"}}},"commit":{"parents":[{"commit":"75051dd5a2cdeef2919ba2062714bde7502416ba","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/75051dd5a2cdeef2919ba2062714bde7502416ba"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-10-21 08:59:05.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-10-21 09:52:53.000000000","tz":0},"subject":"WIP: Use inject metadata properties in upload","message":"WIP: Use inject metadata properties in upload\n\nChange-Id: I500f85b42873239d751031d1714eff989e463f93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8177f388c79f30970b16e315ddfeb133f98005f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8177f388c79f30970b16e315ddfeb133f98005f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
