)]}'
{"id":"openstack%2Fglance~863622","triplet_id":"openstack%2Fglance~stable%2Fyoga~If36e64b61767dc5c1df0daec2dd176fecd409926","project":"openstack/glance","branch":"stable/yoga","topic":"remove-constants-check-stable/yoga","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-11-07 22:13:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If36e64b61767dc5c1df0daec2dd176fecd409926","subject":"[stable-only] Remove glance-code-constants-check","status":"MERGED","created":"2022-11-04 12:05:23.000000000","updated":"2022-11-07 22:16:00.000000000","submitted":"2022-11-07 22:13:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"863622-remove-constants-check-stable/yoga","cherry_pick_of_change":863527,"cherry_pick_of_patch_set":1,"meta_rev_id":"6cd210052308a40e61c392fef263f76057ff4a66","_number":863622,"virtual_id_number":863622,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 22:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-11-07 20:19:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-11-07 20:19:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 13:13:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 14:22:58.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-11-07 20:19:32.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"5f16dda699064352a42df754d1250b1c3e33cb89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-04 12:05:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"1e82196c53ba43f1c4e3e92a413bd946f16fd746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 13:13:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8ff21d09f6348f38d2451138c759dec\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/00ea490df03e4c009fb913631d477227 : SUCCESS in 1h 02m 54s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad506ac8889d4c72934c49a881c5c41a : SUCCESS in 56m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e982d334bb648929723b30ffb6b53a0 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1befb8f6ec344f688caf8b30e4b3b1b : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ef5faeb74ef45f887d0fb797eedf07a : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcde629c2a0c4f41b0e26910415709e0 : SUCCESS in 5m 48s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/7340c58036d547ec822656e38f962d2c : SUCCESS in 11m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e873661a64e64a8690f573d1719dd9ce : SUCCESS in 8m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1b110fa3744147738d3fda2ba623e10a : SUCCESS in 8m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a7ce8016a9f44767a66fd0d1c6a83005 : SUCCESS in 38m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e7d82bfa7d86c57819de04a1fabc25188bba7549","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-11-04 14:22:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"416aa051c846ef2bc4920bc8c76fce670f3b1e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-07 18:06:15.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b24fd47e3ff40226fe8c7d4f98dea16e5208f63f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-07 18:07:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ba4ae15bc3293628b61868f0b74eeaccdc2b42d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 19:10:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/879a74cb7f604da29296ee69acc0a6cb\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/44eeb524ba794e2b843f23819485dd83 : SUCCESS in 49m 50s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21d770d5bcbe441086a66db60a9610af : SUCCESS in 50m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8ae6e851fcd48c6bfda2d3494c2212a : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72193dfd0fcb4049b14ba3dd9765a9f6 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b6b771415294f33a9d3cc5064ca9145 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6e3ad00972640f1b8280667b467bfb7 : SUCCESS in 5m 56s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/d3fc8e9bcc624306b400156b773daa97 : SUCCESS in 10m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc6ede876529421ab3d89f34be4408d9 : SUCCESS in 9m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8c67a856936e48cf9d8ccfae21ac7ca2 : SUCCESS in 8m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/07a9439f8c2449b9bcc6f2e8853905e0 : SUCCESS in 28m 38s","accounts_in_message":[],"_revision_number":2},{"id":"6bbf9061f5fb7bb7d5b0caec00f6c7897ebf27bc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-11-07 20:19:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"308249dbacdef669722bbeb1e3ad1cdb00c27ca7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 20:20:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd8ab70ffd7158127a28ba39c55678eb1e82e156","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:13:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02b20132de754b27a94c00bebfc6aa7d\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef3715df0571497f8d598d3547890d88 : SUCCESS in 59m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1720190f50614c8788080d589943eb9c : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd699c64dba84b59aad62d0d536a8f8f : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bfece6ca1cd4178bde53019f1eac07f : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1cd9ff789e4413acc115b334fc957c : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3324f0169d284bcdb28be076cb9efbfa : SUCCESS in 11m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/40ae37378764495cb50865cdae9743fb : SUCCESS in 8m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74922a3313cc455b8a40a355656be05d : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":2},{"id":"d50b0c61bbc66916b563c2945c39141fb9d8848b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:13:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6cd210052308a40e61c392fef263f76057ff4a66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:16:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ea62cd870414e818a57983481263ca3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f13d40603a341ada316c8e2486487ec : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5fbf47378e70537c73394408e69e94449aba1bec","revisions":{"635d3354efd4a65ef95ff8d5d9002c780fde1cdf":{"kind":"REWORK","_number":1,"created":"2022-11-04 12:05:23.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/22/863622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/863622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/863622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/863622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/863622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/863622/1"}}},"commit":{"parents":[{"commit":"f3ae67fa1925375fef6af44e73db201db4fe359c","subject":"Mock getaddrinfo in cinder store tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3ae67fa1925375fef6af44e73db201db4fe359c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-11-03 14:37:07.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-11-04 12:05:23.000000000","tz":0},"subject":"[stable-only] Remove glance-code-constants-check","message":"[stable-only] Remove glance-code-constants-check\n\nThis job is only useful in the development branch; there\u0027s no\npoint in running it in the stable branches.\n\nChange-Id: If36e64b61767dc5c1df0daec2dd176fecd409926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/635d3354efd4a65ef95ff8d5d9002c780fde1cdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/635d3354efd4a65ef95ff8d5d9002c780fde1cdf"}]},"branch":"refs/heads/stable/yoga"},"5fbf47378e70537c73394408e69e94449aba1bec":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-07 18:06:15.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/22/863622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/863622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/863622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/863622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/863622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/863622/2"}}},"commit":{"parents":[{"commit":"f3ae67fa1925375fef6af44e73db201db4fe359c","subject":"Mock getaddrinfo in cinder store tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3ae67fa1925375fef6af44e73db201db4fe359c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-11-03 14:37:07.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-11-07 18:06:10.000000000","tz":0},"subject":"[stable-only] Remove glance-code-constants-check","message":"[stable-only] Remove glance-code-constants-check\n\nThis job is only useful in the development branch; there\u0027s no\npoint in running it in the stable branches.\n\nChange-Id: If36e64b61767dc5c1df0daec2dd176fecd409926\n(cherry picked from commit 06a9228809d574c4e5b1b722e8238b3c45e58885)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5fbf47378e70537c73394408e69e94449aba1bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5fbf47378e70537c73394408e69e94449aba1bec"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
