)]}'
{"id":"openstack%2Fglance~863926","triplet_id":"openstack%2Fglance~master~Id75b1506ff8cfc873e2bf6cbf51b73e671cd58d4","project":"openstack/glance","branch":"master","topic":"test-cinder-target-thing","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-11-07 21:18:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id75b1506ff8cfc873e2bf6cbf51b73e671cd58d4","subject":"DNM: Test cinder target devstack change","status":"ABANDONED","created":"2022-11-07 19:20:02.000000000","updated":"2022-11-07 21:18:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d5f8c873c6868c1bb80b30e135ce2888a3cbd043","_number":863926,"virtual_id_number":863926,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-07 20:51:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-07 20:51:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5a23d60a2a13f04897a2cd85e8716ee5cbfc9f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-07 19:20:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dff2a6028ba561aab622a76d1a3802c260c47ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 20:51:33.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/21c4162aa87348abb969df562b04cb05\n\n- grenade https://zuul.opendev.org/t/openstack/build/44ce6fcde88740578fea3c4cb675a7cc : SUCCESS in 58m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/338e5da0256d477eb0b3abd4632bdde8 : SUCCESS in 1h 05m 47s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/783b5cebb772494d993fd369b811b9ef : SUCCESS in 1h 24m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e71e757f81684df7a745acd6dfd3d260 : SUCCESS in 42m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2362adbef8a46438b2563ad6f16aeaa : FAILURE in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21c331f8b5994c32a43bcd877ead6bd4 : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58b11589541b40b2bfc41bb21ff98e48 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087db16f267040a7a33bb9e62583b15a : SUCCESS in 6m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5fa5a576b303495d92ead079dcf86125 : SUCCESS in 10m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/325a52a27ffe4205b2841d86ca97bf10 : SUCCESS in 9m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/735218456a894871b3185671ebc4da6e : SUCCESS in 8m 27s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/187be17dd579415cb8b79da17865f6ef : SUCCESS in 4m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/63e68aa0bcac44319938d9be2d27376c : SUCCESS in 1h 02m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f1cdfce289aa4f32bd88b45c5156f418 : SUCCESS in 1h 01m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/33779499f2f64723b641699dd7b5a57c : SUCCESS in 1h 18m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/671901b543764e9c9830468a29941fd5 : SUCCESS in 1h 27m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ea554e6e3574e4ca970051b596c112d : SUCCESS in 41m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/28d32ccee9354ddb94004f671e9d56c1 : SUCCESS in 1h 01m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ab7f646e8ecc47399bc15da17ed98fa3 : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d5f8c873c6868c1bb80b30e135ce2888a3cbd043","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-07 21:18:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"495a6b73236e65c6db86576f50a0d89fc6755087","revisions":{"495a6b73236e65c6db86576f50a0d89fc6755087":{"kind":"REWORK","_number":1,"created":"2022-11-07 19:20:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/863926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/863926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/863926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/863926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/863926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/863926/1"}}},"commit":{"parents":[{"commit":"fbe32cb44b23b183ee6bf63d630de0629020d1ac","subject":"Merge \"RBAC updates: drop system scope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fbe32cb44b23b183ee6bf63d630de0629020d1ac"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-07 19:19:46.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-07 19:19:46.000000000","tz":-480},"subject":"DNM: Test cinder target devstack change","message":"DNM: Test cinder target devstack change\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/814193\nChange-Id: Id75b1506ff8cfc873e2bf6cbf51b73e671cd58d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/495a6b73236e65c6db86576f50a0d89fc6755087"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/495a6b73236e65c6db86576f50a0d89fc6755087"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
