)]}'
{"id":"openstack%2Fglance~863943","triplet_id":"openstack%2Fglance~stable%2Fxena~If36e64b61767dc5c1df0daec2dd176fecd409926","project":"openstack/glance","branch":"stable/xena","topic":"remove-constants-check","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-11-09 10:21:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If36e64b61767dc5c1df0daec2dd176fecd409926","subject":"[stable-only] Remove glance-code-constants-check","status":"MERGED","created":"2022-11-07 21:55:38.000000000","updated":"2022-11-09 10:23:28.000000000","submitted":"2022-11-09 10:21:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863943-remove-constants-check","meta_rev_id":"a2dce9b6c774abff82b98fc19c5b2859bdfb7c9a","_number":863943,"virtual_id_number":863943,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-09 10:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-11-09 09:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-11-09 09:50:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-07 22:27:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-09 09:50:15.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"2441ca8d5f3333c0b1f946f554b784c2778e8ba7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-07 21:55:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"972468087b6605814c3947ba0bacb47bce7dc8df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:27:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcde7d951bbc4ba9ad7bdfb4971afec2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0da3dd7e362f47c5888f1f827143318a : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf931f2975c841f8a9734d6d139904c5 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/133b79948cf947dba314b707a58f4049 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ef8a25acffc4edb8709a40881352aa4 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae4296f35a844f0ae087f4afeffd594 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8fefab516493436eaf1a6025fb3d54b8 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/21eb86d4b8284a91a9b1bae8686390d7 : SUCCESS in 8m 54s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/6af5a07122854464ad3db5f79afd5aed : SUCCESS in 7m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0a7987ac87cd4a699b4f1c5e43dfa608 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4b4f60d556e33c6e47093ee241f10bb3bdc6d302","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-11-09 09:50:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2eda4b5ad67e361349ebf36f7ed79e20e8afbaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 09:50:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1cf03a7d0092ff58316ff046bf7a6fee8066a40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:21:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e44ea4d280194bfb85b29062b49e7f6f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e7848d66e44238a792d819609ae27c : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1947d15e9d44f3ab2d9b687c2542905 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75531f0125e04275a590cb69ab54de3d : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f30ee7f054e44ee292640a54a8c4e035 : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/187869f6dc8346a18cffc42c2825ffeb : SUCCESS in 8m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d4ff9bf343724a5e879f62431f153483 : SUCCESS in 7m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/136c78aaf68b4d3084c83ffb50d6b515 : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5e20c3f2cc75bfbcb79705bfe869194e74ed2649","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:21:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a2dce9b6c774abff82b98fc19c5b2859bdfb7c9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:23:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a68a99064f4147ef8a4cf17e0105ada4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2206e393e6a40ecb2af58376e13389c : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3cfbb69c017ed0e2c61f267829a3b79b073832f0","revisions":{"3cfbb69c017ed0e2c61f267829a3b79b073832f0":{"kind":"REWORK","_number":1,"created":"2022-11-07 21:55:38.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/43/863943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/863943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/863943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/863943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/863943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/863943/1"}}},"commit":{"parents":[{"commit":"cbe6522e67c2e437e600b4fb153a1d048a310282","subject":"Merge \"glance-cache-prefetcher: Set up access to multistore backends\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbe6522e67c2e437e600b4fb153a1d048a310282"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-11-03 14:37:07.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-11-07 21:53:41.000000000","tz":-300},"subject":"[stable-only] Remove glance-code-constants-check","message":"[stable-only] Remove glance-code-constants-check\n\nThis job is only useful in the development branch; there\u0027s no\npoint in running it in the stable branches.\n\nChange-Id: If36e64b61767dc5c1df0daec2dd176fecd409926\n(cherry picked from commit 06a9228809d574c4e5b1b722e8238b3c45e58885)\n(cherry picked from commit 5fbf47378e70537c73394408e69e94449aba1bec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3cfbb69c017ed0e2c61f267829a3b79b073832f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3cfbb69c017ed0e2c61f267829a3b79b073832f0"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
