)]}'
{"id":"openstack%2Fglance~863947","triplet_id":"openstack%2Fglance~stable%2Ftrain~If36e64b61767dc5c1df0daec2dd176fecd409926","project":"openstack/glance","branch":"stable/train","topic":"remove-constants-check-t","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-03-03 00:07:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If36e64b61767dc5c1df0daec2dd176fecd409926","subject":"[stable-only] Remove glance-code-constants-check","status":"MERGED","created":"2022-11-07 22:15:47.000000000","updated":"2023-03-03 00:09:48.000000000","submitted":"2023-03-03 00:07:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863947-remove-constants-check-t","meta_rev_id":"6e276e5ee40517666ff6dd456ab51d63195fb11b","_number":863947,"virtual_id_number":863947,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-03 00:07:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-01-25 06:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-12-15 14:45:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-01-25 06:33:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-07 23:21:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-15 14:45:54.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-01-25 06:33:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-01-25 12:46:26.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"fdaa942dcb59bed21642d4770489ed4ef4b2240d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-07 22:15:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa978c5fc09e2cebebd7266a9f878868143671fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 23:21:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3f6aaef103d4f5e8cbc3c7ba077fa33\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c498a6b82a6e4fd3b546c296b1e3e0af : SUCCESS in 1h 00m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/629b181ab3ac4feba22f053f49ced92a : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a8f2ecd5e8b4c86ad14b59c4afdc621 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bae003d5685f4734a4bfad837320347e : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3ee4e561007482a9c99dfcb0649b962 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e590df5fd843bf81de2d790294c287 : SUCCESS in 6m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c5a84bfcbb504559afc39ad99cb0e41a : SUCCESS in 8m 02s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a7d21830d01341108eff83e01202e57c : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3f1a4720af44170e8dd608375663b937fa8efd9d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-12-15 14:45:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b4e04e68e7803df1b7d87f7aa8b8edae36373505","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-01-25 06:33:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0509642ec23697a1487e51ba2ce818217a2654eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 06:34:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"278db520d2f2a4ebd3ae0d01dd5f208f7e833652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 07:43:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c84313b3c8964cd98021fb14e0a644e9\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/97e7493b14544acfa8d19a18dd683b41 : FAILURE in 58m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fde5bc749224bcdb9bc8101138736f0 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ca15c81042c141b98c78e96dc2326ad5 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f65f4402b2541e0901569aa88eeb760 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33ebc40326c64262ac52f24f99bbb3ab : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c7e4285ddd4907b46125bf7ae0e36e : SUCCESS in 5m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cd524d4fb94f498c83e234635ff40d25 : SUCCESS in 8m 28s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0edd799ce93247d3b77dacde2a85d7ce : SUCCESS in 9m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ef9e1a704e7afa9271e2809c26d383d8174819e3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-25 12:46:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2fc1a94e7d4d2ee33d470cd184598e2d2f68114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 13:49:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cad034e489e4e149a09da7123c59a17\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e14c167aaae04720aab21e29b4f6f064 : FAILURE in 53m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/521bf0d4ffbd4ae2b872ad594d5c64f2 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c74102b62b494b4cb4c0c081277076a2 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48d01b311a844c68a87e31c259a17d33 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6046a304a78643868f2c5d6d6f517f3c : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80542263db0848c7b49d614e161be69a : SUCCESS in 6m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/26474bffcd5645d79f9815f1e02f88b6 : SUCCESS in 8m 20s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0b0ea95a402d40c4b2c8adb539158558 : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1d0ae5e74200360051a5ae5b4daae4e5f25f7134","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-01 12:35:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"201c8c6eef6d0360b564da4abe0b90674cd4be8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 13:33:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78655981f5c64faab4a5bc71689bf12c\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f9757e5179674225ae13623f6d960967 : FAILURE in 46m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d90be52ce16e4d379421164c47102a0e : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00538fa66e9f4823977da38c7b166c06 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8f7d2e4195e429699bfa1b73100c977 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4f9bc22bbd140f98c4efd5cd5f679b5 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce1f48a830e46acb06818e0d0633143 : SUCCESS in 5m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0e6fe75e37524ee389521c979cda32ee : SUCCESS in 9m 16s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d33ce51589f94fb1bdc2e7abaf10eac2 : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9c88aa44351e27e9b3d4b46acf5a47beffd0d2fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-09 14:31:19.000000000","message":"Abandoned\n\nThis branch is too old, we will not merge this.","accounts_in_message":[],"_revision_number":1},{"id":"8a5921f638ecad2683a1c0d0d59df7546a70721f","tag":"autogenerated:gerrit:restore","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-09 14:32:12.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"682ac328534d2987f50ee252ee6230d3d2dd9e69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 16:41:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af24da8e9e23458caff3a6f3f04a17c4\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2a995706ad934058955f0b2499ad7eef : FAILURE in 48m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d0ce8f0a6cd4708899faf08085a9d33 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a868dc553c85439ea1942cbb4d09517f : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3092f1f85bd24fad946caee1646576b9 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ddf0ccb4c93f4d0ea7bbbad0142fd448 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0710fd350d4077ae655c6f9de0199c : SUCCESS in 6m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2d0230fa9459400e8768bd0f43477e32 : FAILURE in 3m 04s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d2b28b6507d14f919d7bbd4eb6f24ba9 : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":1},{"id":"13074bc0078a335f480d9c1df34d2dcf8272e7ec","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-22 14:34:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d2a020d07b5a7c86fefd41f4149873bbacd7cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 15:43:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fb62a2c7ed84513b4b0495f51468a28\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/29f0655ddef4407e8f4f94219cf4bd2c : SUCCESS in 1h 01m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/243fd4a7621e4675b4bbe0bc29b73635 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e947cfe4fb8a42e5b6147935e5b108cf : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f982f497269425fbe32f45b5d90985a : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5db2492d862c421482c4253f45fecdd5 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/899487a4564a4e86af1d8acd0d9c529b : SUCCESS in 5m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7936a20b0dd04bb993bcb969c6f96eb4 : FAILURE in 2m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b6ebe924fb4b4dbebfa31865615afc76 : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2856155926ee0498bc20b084ee03e82ed8fe0183","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-23 06:02:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d49247e4661abb43d9356efeceddc671eae440a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 06:58:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99ccc48e02e84a479d83913ae21d1584\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/26020688e0454ddcbe04c64619447791 : SUCCESS in 54m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6c9bfdb77fd4ae5b0719f46062b6a1a : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/beffe22671734257a4aed972f98ba523 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c82de60449b4aa1a7f06418c065c565 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73eedb8a8af54b27b78c3e291325ad2b : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60812c40136d46e291eb53ea05ec2d6a : SUCCESS in 5m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4080afbe860c4a5b8e1225f4ff45ccda : FAILURE in 2m 50s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/fe615e9f55f24512a3bff12614b324b2 : SUCCESS in 8m 25s","accounts_in_message":[],"_revision_number":1},{"id":"816565759db3bed57df9aa994b1bb748fda5ee3c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-03-02 20:14:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ef51ebc57bcb6e15502f83bc62a9b8e4f79517c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 21:42:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/830870be136944f4af9632cf9062a426\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d7733f6e5fea4c7ba1fda7199b5b33b3 : SUCCESS in 1h 03m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30e7054e08e54266bda4b57291270dc8 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/810f1cceefc44f0e8b0bd4e624d4bf28 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9e8321fc0174fcbbe7b192dda8ebfed : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc7aa5e7012c485b9d00b9a5e7837c01 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95167250d85f44229f10a59882d6784a : SUCCESS in 6m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/da0c1e73d72846afabe612ddd595908c : SUCCESS in 10m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c9726bccc4b64a4184f6574ea3e3ce92 : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":1},{"id":"09a1d4c675de0694b0c0afa43168a179d427f1af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 21:49:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"676f0833d84ba980c820c9dc888ed2af7aafaa22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 00:07:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09ef0617cd944c02b60fc1877cc94454\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a01e0cd145164a549c380374b54d44cb : SUCCESS in 45m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b982652991ab4633b5f8495883211eac : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4eb508c0c67541e1b0f15be29dae8d3e : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d422b36adb7546a6a06cfb03ee8a0b7a : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ed423d5bf2a4c18aeb0a5aa671b7d67 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e46c1ecb95de4bb78200d855b9e8b660 : SUCCESS in 5m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/81ef435246ef4931a8cdc47b3d9b1701 : SUCCESS in 8m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2a53d5f2466a422aa4632bcee50c092a : SUCCESS in 9m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4e25982170f63a3143e7521857def3abec03d63f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 00:07:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6e276e5ee40517666ff6dd456ab51d63195fb11b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 00:09:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9737cb0d17d843fcba4426efb67d5921\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18060951a554da389f9ec60ed627705 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5930c56125886710443321bc4f863f8b2e81c31a","revisions":{"5930c56125886710443321bc4f863f8b2e81c31a":{"kind":"REWORK","_number":1,"created":"2022-11-07 22:15:47.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/47/863947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/863947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/863947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/863947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/863947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/863947/1"}}},"commit":{"parents":[{"commit":"b3de9da3b9c0f32cd0c5f1eda6bf1789b7425dc9","subject":"Drop lower-constraints jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3de9da3b9c0f32cd0c5f1eda6bf1789b7425dc9"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-11-03 14:37:07.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-11-07 22:14:33.000000000","tz":-300},"subject":"[stable-only] Remove glance-code-constants-check","message":"[stable-only] Remove glance-code-constants-check\n\nThis job is only useful in the development branch; there\u0027s no\npoint in running it in the stable branches.\n\nChange-Id: If36e64b61767dc5c1df0daec2dd176fecd409926\n(cherry picked from commit 06a9228809d574c4e5b1b722e8238b3c45e58885)\n(cherry picked from commit 5fbf47378e70537c73394408e69e94449aba1bec)\n(cherry picked from commit 3cfbb69c017ed0e2c61f267829a3b79b073832f0)\n(cherry picked from commit 1471384ce93a3abacd02d725eb2938d160f000d2)\n(cherry picked from commit c9a0215e3527e0daad9ff16277811414eab2e07b)\n(cherry picked from commit 7ffbbebd8815893b47958908be79cddc1a677ba0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5930c56125886710443321bc4f863f8b2e81c31a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5930c56125886710443321bc4f863f8b2e81c31a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
