)]}'
{"id":"openstack%2Fglance~863977","triplet_id":"openstack%2Fglance~master~Ib7025845656664bbbc44b4b7a160e6dfba4e783a","project":"openstack/glance","branch":"master","topic":"863977","attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-03-24 07:46:16.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib7025845656664bbbc44b4b7a160e6dfba4e783a","subject":"DNM: Test cinder store extend attached volumes","status":"NEW","created":"2022-11-08 09:36:12.000000000","updated":"2023-03-24 07:46:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5cd3e5fd85f250bf71deb30406569d9e87db89f7","_number":863977,"virtual_id_number":863977,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-03 15:52:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 09:36:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-24 07:46:17.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"624ceb7a2184453c6b78f1859e1b271d01f14b2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-08 09:36:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf6970a905659e021c7790f560b00af6c62bd4e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 09:36:39.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/glance/.zuul.yaml@master\", line 369, column 31","accounts_in_message":[],"_revision_number":1},{"id":"f7e6ca09aa0dc89fc50fdef433d1fb9822493a73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-08 11:43:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e86656e4ad1009e43557078b663ea9c0b635b070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 12:51:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4f3912f35d54076a6d4dc7f8d63e745\n\n- grenade https://zuul.opendev.org/t/openstack/build/4b87c75c13ca4425bc4ad9a879ed5731 : SUCCESS in 41m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a25dd5852fb94fed9d37c24e1879326d : SUCCESS in 39m 02s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/146ea8dfb4334519865489961b1af4f3 : SUCCESS in 1h 06m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d539c638858146d4a6d390f69924c33a : SUCCESS in 40m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3ae5a4648b496a8579ebaf35d2e74d : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/473dddde4d374b03bb10aad413eacfb7 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c33e912b2ab4432192652f2a3d2838cd : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed52fa3d198b4376b57667e552a114ad : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"a3e8e28f08e0dd8f366e05e3b321738e11befd34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-11 08:30:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"be219476b67506cf6f66bb277b3ea6094106cb1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-11 08:34:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8a174e3d026fbfc3e517142fa5dd6031263ba9c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 10:07:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c362eee696541ca829cdb4d7992cc0d\n\n- grenade https://zuul.opendev.org/t/openstack/build/b84fd7b8ad514be48a05c69afe14550b : SUCCESS in 45m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f40e265705154eb88a6bc22d5da9474d : SUCCESS in 1h 00m 25s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0de0447c0f71441f8117f03989606787 : SUCCESS in 1h 27m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2609fb8155e74d3dbbde962998818db2 : SUCCESS in 40m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/227c2eb19bf343eea9616de6a0b1f6b4 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc1f620aa0a5474e8546ba17561fd507 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e9a0a8e55794be58340c2fae85ac5ee : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80085c86db31471eabfaf1b15e830a02 : SUCCESS in 5m 47s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/3fb676c23b104a1fb3c1c32c1193be0e : SUCCESS in 1h 21m 49s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/6b79dd0714cd4a2cb269ee0e4d47c173 : FAILURE in 22m 03s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/d3dcd770b4ff4348849861eb1023537b : FAILURE in 22m 34s","accounts_in_message":[],"_revision_number":4},{"id":"6c62df7eae00ccbbcbbb74fd46e25fdf63ea3525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-15 08:18:49.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"d75b5d4216d24ab35413d4ba88fd0e5453a73cef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-15 08:32:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"01158b93778142fce5281bd4c4bd99ef4e33c3c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 10:18:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dfa756e6a4d4bbdace1e0376b3a3604\n\n- grenade https://zuul.opendev.org/t/openstack/build/fc1b4686d6204bab8ea8d13ea88556d6 : SUCCESS in 1h 02m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d3d8e6b54a3640d8ba3768442c8b1292 : SUCCESS in 1h 05m 09s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a5eba6dffe644c4b0db10cb259ee263 : SUCCESS in 1h 25m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddf10dad9aef45beb07532f01ff63f9d : SUCCESS in 57m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51637abca4c34a93ab83f68f4323dbd6 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b64610924a4f43c99cff260cb7f450f8 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3800059a62f446fcba074f373eca9d7c : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c01f6475b0475a9744336a73b9840c : SUCCESS in 6m 28s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/90f82c91797b4a809634a45dfb4bc246 : SUCCESS in 1h 40m 35s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/daf1147efb4e427891fa5461358e6d13 : FAILURE in 31m 32s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/8730e0640dbe4ca9a61af64aa404709f : FAILURE in 31m 49s","accounts_in_message":[],"_revision_number":6},{"id":"5c58f8fa9cafdaf7bd6188efa01b54f34c32d981","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-26 06:28:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0bf87a19a82ec3c9f1b74926efaf2d52c74b4a53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 06:38:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4699ad88b1b4700ac5d1bf14ed539f2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d3bea7a81354ab09e5b8aa5f3fabf19 : FAILURE in 7m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c550ebc50074ba29de9b6c9ed994751 : FAILURE in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8cdc79ec7f0b7511b2d0b97a3618e580b2a7424e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 08:07:45.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebf3303bc2aa4ce699a207e614e3c3bc\n\n- grenade https://zuul.opendev.org/t/openstack/build/b01212deba1541bc9994888202a9f947 : SUCCESS in 1h 01m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cbc692e5005e40299da6353017531fa2 : SUCCESS in 52m 05s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2bc15892ad4440848bc27d1c66d3f503 : SUCCESS in 1h 34m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aab733393ecc4004b5f82346b1b9f861 : FAILURE in 31m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ad912e57f14e5ea372a2d9e17e9a38 : FAILURE in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14f35b97fe7b43ec8dc9510be5ad2c55 : FAILURE in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee73999e743a4733a2f6e045d7efe3a9 : FAILURE in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f17f6a6bb0654f5683efe5597c8f53d4 : FAILURE in 3m 38s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/9c57d12937d54bd5895e134d4419ff31 : SUCCESS in 1h 20m 26s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/0f821b107d634dd6a571a48276f8ec20 : FAILURE in 20m 04s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/be95099c8d7f450d9f5c06b5cb353355 : FAILURE in 16m 48s","accounts_in_message":[],"_revision_number":6},{"id":"f07138c0e7504da62e11ffce259fde1684af0b90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-26 08:49:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2c3c204528b6c0e38273f42f15ca20b3e8e1aacc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 08:59:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45a88e62c46c4ce99ab93b988da3fcf5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/16b66f8431954239864ce06f2cb5b3d8 : FAILURE in 6m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7518c9add56b45eb961ade6e508d1e8c : FAILURE in 7m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"45e985a472bff1bad912f4f54e5d22bf77453277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-26 09:50:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1e3f78335954f8d87bf8f59a320741dd3cf8b0f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 09:59:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f578c18385a543f88156bd11f0a07b96\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3318550ba8a649e18c468e6764e0208a : FAILURE in 5m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/471512ded607400f97dc12ff27898b62 : FAILURE in 7m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"93208d5cefabdfa232bd85ee4957af21a2bbe3c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-26 10:07:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"33aa92aff9d43ed2653d938daac2f20a0af53c4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 10:15:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42e47b2562e5448ab13129ca15c354c2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf77e7bd08394985bf12ae8242b42911 : FAILURE in 6m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2585c074ebf44cdb9fc0aee57f65ebfa : FAILURE in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"aa04a5ad53eaa4f9c11cb0d0532a9d0e54f28846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 11:52:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb5e8d533dc94bb9bfd6a754803f70d2\n\n- grenade https://zuul.opendev.org/t/openstack/build/43303ae5a94c4206ba3b7b55ab1c26b3 : SUCCESS in 58m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ac5b81a6d3d54797bf0d54c6c93800de : SUCCESS in 1h 04m 12s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91db13cf97cd4ff38ed9c86da4ba64a0 : SUCCESS in 1h 39m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0c79ee12279416caac23fb19521da66 : FAILURE in 17m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/209cc7a92ef249e8a098663929036564 : FAILURE in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/882b054e40ba49139a9237271ff9133a : FAILURE in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/433cfe7769f94b43bded8a47118afb40 : FAILURE in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a6927a971843f18de30357e0cfc9a6 : FAILURE in 2m 50s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/aa32b606762e49a38548d80b381aae22 : SUCCESS in 1h 05m 17s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/5ca84fdcb6da4a41ad25df68c26c86dd : FAILURE in 29m 24s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/50837462b7b0413893488807ffd624e7 : FAILURE in 18m 46s","accounts_in_message":[],"_revision_number":9},{"id":"1669eb1ed1e2a4a7919443f776161371501f4d0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-26 16:28:16.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"57253c867946f9ff7648b0827f5b00410bfdfc5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 16:37:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9920384d1a774216a9148b2021172d38\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13c63126b5f44b60b0f7b2365f6f1ddb : FAILURE in 6m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da7c114985424b668e8e8e02058f87ea : FAILURE in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3799a83569453ffe4b99bbb5ea0a53ca1fe7b137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 18:21:57.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b77ee4bd5144490bbb9e4395b4372ede\n\n- grenade https://zuul.opendev.org/t/openstack/build/64067e16158440fab6dc09d1ee0df5cd : SUCCESS in 39m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/de3c0e8b7a8f4f1bb580892220cc99a6 : SUCCESS in 43m 18s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec57d2b436634f9e8949a96517ad9dd1 : SUCCESS in 1h 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d53401a5adee4efd9e081203ea6b5dfc : FAILURE in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81141de8014b490385ddf45a913425a5 : FAILURE in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c085c5e9240499a9beca849f8a5f752 : FAILURE in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f0544ecce7747858e4a60763c251537 : FAILURE in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b8d4c1b96544589611ebb06e9d192c : FAILURE in 3m 09s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/8ac9d54c8445475297442f2461d5d000 : SUCCESS in 1h 48m 26s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/540268be806c443d94a39bc8bbd34b64 : FAILURE in 17m 24s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/914fc8548de64624b4b61e495197135c : FAILURE in 32m 20s","accounts_in_message":[],"_revision_number":10},{"id":"9895256e5bf761d7c15b65fdc9fab82c8815ffbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-27 06:18:00.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"40c1ddfc0295896b25e235c47add1f78f5c1078a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 06:27:06.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00a143059a6044df8be7e6d45e160758\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/742ff8cc38034e2598fff1e9246cf889 : FAILURE in 6m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b54e8e05d19843aba298b415ea100ef4 : FAILURE in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b947b27393cd73fb057cda3e55f8a41653008179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 08:15:48.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93c6c52e47564cdbb393320f7c3a60ce\n\n- grenade https://zuul.opendev.org/t/openstack/build/1cdb74cb23c64653a792b9c71563b9e2 : SUCCESS in 42m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8859372ae4234aa48ebec6abb30cf5a5 : SUCCESS in 44m 10s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cfe606b6e13041bda2b004bf21165219 : SUCCESS in 1h 16m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/231a07def1094624aa70975a3389f792 : SUCCESS in 40m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ea3014576ea4da88c7d90633ab78e93 : FAILURE in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/893de7a463284df49ae73ad242eb5427 : FAILURE in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3533552e748c40fe927f96c40e46554b : FAILURE in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088d70bb974b4ba7a69c4245e57e97d6 : FAILURE in 2m 57s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/1350a76a09514b7797e08843d3c92110 : SUCCESS in 1h 52m 15s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/b0c454588c6846a59f26f4915dc6268c : FAILURE in 17m 27s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/b8b0603f362d42c08061a1d3d9bcc042 : FAILURE in 18m 58s","accounts_in_message":[],"_revision_number":11},{"id":"fa1758ceb31b4bd14b403f7fbbd8395dc5880a7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-27 11:53:57.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"61e1ed0881041b754cdb1f80a8d5bed9e54140f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 12:05:28.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3661ef0232da489e95b42de5e675b439\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38f245994f554ee184622a613c6fc093 : FAILURE in 8m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/46d0fd7879df401181ac51247df509c3 : FAILURE in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"97f1051524cf95e90a8f0bcd078ec84c0fd2500d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 13:06:09.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58762e7ee7d346bfb654852b5aba7206\n\n- grenade https://zuul.opendev.org/t/openstack/build/dfe4701ab7a644b5a5d37611f4c1fbfc : SUCCESS in 45m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fcef3cb047264d50915d5ea4f8249fe0 : SUCCESS in 42m 41s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4a812c22b7f4da98e606365d9487959 : SUCCESS in 1h 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06e6ef07335a4ec48e3ce53605748b2d : SUCCESS in 41m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d4537a3cfd4234ab3e07a711085df7 : FAILURE in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/530d7d1155d6468d93091b05c0823702 : FAILURE in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a005c7622814be382d32d631d6f177f : FAILURE in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0831c571c6554041b7808d160ab82f1e : FAILURE in 3m 06s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/e88b40285e9f4dfdb0b194de8d3dec8e : POST_FAILURE in 15m 41s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/1ca51317051047e4975753724e9adee0 : POST_FAILURE in 16m 09s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/6b95132b3ec9447588ac0ff9e4a96d3b : POST_FAILURE in 19m 39s","accounts_in_message":[],"_revision_number":12},{"id":"eff22939296e05a8d0be1683c6b2113ba1f0916b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-27 14:34:12.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"2c7a2c285c3c3024d0d92824ecccf8e967a4ba66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 14:45:22.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77a98160b1e14170bb420800cdba9fba\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4591ff393fe347ae9cd04c908787607b : FAILURE in 8m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d4bee65ac0b34cafbe6ab92270325e47 : FAILURE in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e7813fc31e3385dd353c8374b0643b08a5427136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 15:40:51.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5982fb743da048369fbefe99d355fc71\n\n- grenade https://zuul.opendev.org/t/openstack/build/41f2246d2e1f49df94aaf512971a44f3 : SUCCESS in 1h 01m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3e97f7c974a4414299193690e49ed324 : SUCCESS in 47m 25s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dba88f0b1c0e47afba587b7286bb2665 : SUCCESS in 1h 04m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c65deb322e544d8911cff1072087050 : SUCCESS in 38m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23f3f2c35c3144a1ba843bd3004cfbad : FAILURE in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d07b19059310438291a7f56c85e198af : FAILURE in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c83b3930b6e4676b3843bf6d3e9adc3 : FAILURE in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12da58608ff64126b61d90a9bcca2bbb : FAILURE in 3m 49s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/2a010cb09f7a4d11b04f6b33741705d5 : FAILURE in 21m 08s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/0fc704327dbd4c0f95752950b44abd23 : FAILURE in 21m 22s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/ed33c974e29c41eba4ff8d6cef450291 : FAILURE in 22m 49s","accounts_in_message":[],"_revision_number":12},{"id":"a1808c740396a74c403b27dc9387f1d6cd648346","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-27 16:21:07.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"445f3a5d125618634997a7a6a5149eb1b02a0e8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 16:29:52.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f74be9b2ef8467990d33429dd174408\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3e5e110503240589668721fa85c3012 : FAILURE in 6m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/862aa44c423a4e7e910763c0518c380a : FAILURE in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"941d658a01a2e6f0c8b77c115fdca0275f75594e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 17:03:42.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/843103 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"3e31ebee029e80053b76cdb8ab39df2a533d399f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-27 17:07:39.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"81339bc8cbb04134fad9b3cc0a73268235806f5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 17:18:13.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13639ba06edb4a2ab70ae35d2a382a53\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25c29875763b49aa83098e15cacaa790 : FAILURE in 7m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dbc1439089dc4f00987731bf929e7701 : FAILURE in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d5d00a07a32c0102d1b3dd55c821b034068fa8da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 17:35:12.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/843103 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"21571738535bf7a578ee29402282b8582d9f682e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-27 17:35:38.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3604dea30f5cf893ae41584f710645f30197daf4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 17:46:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15c17d03f9ed410b9fd2860a92962d84\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2cbed7d75a548d4ad77eeb46c82fb54 : FAILURE in 6m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f9c2b9beb294c7b8ea3eface2b9be68 : FAILURE in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3e7b263815295962e14f59878f0d7513a202cdf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 19:32:34.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fff34104b6040a79e8bd586f33f9cf7\n\n- grenade https://zuul.opendev.org/t/openstack/build/39896a5454894470afea976969c9ae9f : SUCCESS in 1h 17m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2dc9f6d309bd472cbf44eb025001c052 : SUCCESS in 1h 19m 43s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5197aafebc445e2b909c1d973312c3b : SUCCESS in 1h 50m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff01a0e31828400fbf8e5533bc142368 : SUCCESS in 1h 01m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9022f6ce91484447b2c73ca3a46f96bc : FAILURE in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/058b5f73ceb9423f97c8d18b1ac7cc25 : FAILURE in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60ae6c6fa047467fadc15da5455b911f : FAILURE in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3914391066f8419fbb54a86f775d0f9e : FAILURE in 3m 07s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/1b8091edec794278b136020e40b26b6e : SUCCESS in 1h 47m 24s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/06b688757d8d421dadeac1e1121021d2 : FAILURE in 1h 19m 33s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/101c088742b14b48b70dc8b1bfd4281d : FAILURE in 36m 02s","accounts_in_message":[],"_revision_number":12},{"id":"02c38fbd9095717b606687cc1beaab4ea14940c3","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-27 19:52:51.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"6b5075f197d3aba9d180c82d39dc85ffd0d167b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 20:01:18.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26946c978f744124a1d14ed81418dd4b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/835bad2c0ab64f53a421e350928f5f4c : FAILURE in 5m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a5d0770521a74478b9b042adecf169f0 : FAILURE in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"18357e971e9b986a6d1cc7514d2450f11386db73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 21:29:50.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0e339b985a9418a914fdc169b528f25\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff6b7bb1dfff4efcabccac406857ec6e : SUCCESS in 58m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7fc3a48eac4a4509b1b9a26ecf58c090 : SUCCESS in 50m 29s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b1f7973157745868a37943e04e6aaf0 : SUCCESS in 1h 17m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c559e22a738491cb145a7c2eaecd0ef : SUCCESS in 55m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/199fa79bbc4b44fcb05db64d4e3bdc63 : FAILURE in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb92c9f75ccf47cfa2768bced3dd1e5b : FAILURE in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/feed8c0546fd44aa9a17266af7385752 : FAILURE in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0a2c3ca45f4e25938bbabb7b73fd95 : FAILURE in 3m 59s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/ddfd62770ae64b83917d50ae98b4521f : SUCCESS in 1h 34m 51s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/ec93d7d5f32a4ea8a3f87fe785a29487 : FAILURE in 1h 16m 15s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/4e6243e9a0fc461b8a1bd40362b69bdf : FAILURE in 22m 47s","accounts_in_message":[],"_revision_number":12},{"id":"90bdcd9183476f4e33b6e1e06a951a57f0c3f9c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-28 04:19:59.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"5b897bc40b2e62d8dadc86bfb886e77733b43072","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 04:36:13.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b3c9f66afeb480c9dd894ed3a9c48b4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2203f9bea97439f90fe146badf671a2 : FAILURE in 11m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ca1836f4e8549c58425d87a4cd5552c : FAILURE in 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cd622e6fe0577d78ebe68d48f47e8549c2492b8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 06:04:33.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/547414a091824174a766ca802bf312f4\n\n- grenade https://zuul.opendev.org/t/openstack/build/128220a936884937892cebd08d472d54 : SUCCESS in 38m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/382c9991de3346d1acc307278a50545f : SUCCESS in 39m 06s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b8887c83e6b4110954500c685e398a0 : SUCCESS in 57m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8439220db6e84d4195029b8e20001813 : SUCCESS in 55m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b906ee8e2d334cc2bfa68fe72f25116b : FAILURE in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d0467f7ccbc43aa8dd6e0053601fe5d : FAILURE in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6859cd63e13446484529406ca5a823c : FAILURE in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f9933506bf749069f2936d6a6013f54 : FAILURE in 3m 08s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/e69481564b594f1daf20e23549918aec : SUCCESS in 1h 39m 24s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/4fb9230d1aec4e828ad19a2bedb28709 : SUCCESS in 57m 08s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/5e48b844bffe4d329d6716ef1fa6be1c : FAILURE in 33m 22s","accounts_in_message":[],"_revision_number":13},{"id":"a26e57b3960986a2018830a3a9bd687de4425a04","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-28 06:34:38.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"deeb2c23a322c0dc6090b62cea1d025c79773ca8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 06:54:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/449ed225817f4afea3d23a9df6681a4f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5202c044834548ce9dc11824ea616af5 : SUCCESS in 17m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad06574ec2d64508affe1d9ddeae36f1 : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9782f5bf44c0eb4ecc7efed9d24bee88f7ea7ab0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-28 07:49:29.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5dda75cf2bfa41ff9f95e7cb885c719380632396","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 08:07:02.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a88cd2015ac5455083dd06bb17ee0fb8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7006de3d59da451093598331265da8fa : SUCCESS in 12m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3369c252d83487492b327d8e63513e8 : SUCCESS in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ad63c1c1084bf86e2f512843657d3a0b8dfe0c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-28 08:39:23.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"1cb886870f595783c2f00925e6c1463aaf4aa201","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 08:55:34.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/918a1ea7df664a5f81d091e023a8cb20\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1481e658ab204cfcbc63ace9fc6cbf57 : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d0d8ecfc10524914abdf4c88f2e4414b : SUCCESS in 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"370278deeb0a5ffc9e56b524c27f484b4249cd18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 10:12:29.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e586ea2348cc4ac9bef55628a766290c\n\n- grenade https://zuul.opendev.org/t/openstack/build/9c29df37f89441af9cb8c74722477958 : SUCCESS in 1h 00m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5341eca613904513af4e49a362fcfc45 : SUCCESS in 52m 10s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e87841403a1e4792a97adeca969fb8a0 : SUCCESS in 1h 17m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff605274bb5c45889864a282f0125edc : SUCCESS in 59m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cd0dfb161c843ae83f79bf6fa076ef7 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c37b1840e2d94d319d7bae32a090ce4a : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fb754ce781542c6b9fd510c81f0ac7d : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b6d14392e64405bdb800a15bdcd5bc : SUCCESS in 7m 16s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/ffdf44d9a4f34ad8ad124ed055e576bf : SUCCESS in 1h 30m 41s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/0f5a57614a634018807934025329f694 : FAILURE in 1h 22m 58s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/a216e9f0f779407c8c91cf08cc4996ff : FAILURE in 1h 07m 35s","accounts_in_message":[],"_revision_number":15},{"id":"e89f1c15e80fd6e711d2a22e6f5308eda5f72a40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-28 10:46:52.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"b73ae23b6edb85cac9f326800d5ea48557c688da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 11:06:16.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/410a8e53ca4c4f148d7c9946ee86d98d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4560c58d26834713b12761323266c474 : SUCCESS in 16m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9fa679d68f2547f08c6b3e8fbc21b348 : SUCCESS in 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"31ef154be8e1e9c73d2af5ba2701c7b210cc5e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 12:23:13.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d26875ca74ee4c8fbdaaa2e226661ae0\n\n- grenade https://zuul.opendev.org/t/openstack/build/19e12266e2e34b508871ea75af1b1005 : SUCCESS in 52m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/567984550865448481c60edb6179832d : SUCCESS in 53m 34s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e0883468e9f04d039f6b96e268b2da79 : SUCCESS in 1h 23m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdcf69c770aa44c1b308e47c6505edaf : SUCCESS in 52m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27b34e11775e41b79b0e65a4f1d70b2e : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c08394f1f3c47c282e4f8f5699453be : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9fd8f711de24d1dbd9a9d755848cb99 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7097e860c764f1ba1d014c77d70aec3 : SUCCESS in 8m 22s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/22037e898d6e43518a74ebf9058ec71c : SUCCESS in 1h 34m 13s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/a5645e2744b44ddab41597a410284df0 : FAILURE in 1h 08m 43s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/f759ae4e4baf4df5b9407a3e8ca3c013 : POST_FAILURE in 22m 13s","accounts_in_message":[],"_revision_number":16},{"id":"0eda36cfd23fbdb590c3dbf93dc9eced479fe8ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-28 15:07:11.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"f388794e4f5a0d392287ec6d5bcb365cbf104436","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 15:24:40.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4810e48a177431dbbbbbeb1ee6de19c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e2bdc020a6f41f3a27eff8db8a8f2ea : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ef7f4c9b85e440429365883b99b03a12 : SUCCESS in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"818d123589d3a878ad05283d66d5764bceb27c79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 16:35:52.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9f2de07f7944c66ad1ce9121e4c6659\n\n- grenade https://zuul.opendev.org/t/openstack/build/6b9dd23922494cdf9b534ba14648eba5 : SUCCESS in 49m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8a460b6c8bea4cec9c72195cd7584c64 : SUCCESS in 47m 35s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df5a975133fa42bc91cb89c356c53e13 : SUCCESS in 1h 18m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/489fdff467694cf099c34149cb429162 : SUCCESS in 44m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c020d4092854bd1861a63b1e6ed3760 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11f10b757a1a408ea3cf12041f3baafa : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59358e72a85b47b5afc26f06b0bb3731 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/affa3558c0794e47898b3844338774ff : SUCCESS in 7m 32s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/3c84fcd5893f424faafa07643a7b0a64 : SUCCESS in 1h 26m 38s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/38f97dd0db034be9a05fbbf02cfb0014 : FAILURE in 1h 06m 13s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/23158ad1a47f495687b8c4473ee2d800 : FAILURE in 23m 00s","accounts_in_message":[],"_revision_number":17},{"id":"c8218f64b26b198996abbcf632bb92c9b911cb74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-28 16:44:00.000000000","message":"Patch Set 18: Commit message was updated.","accounts_in_message":[],"_revision_number":18},{"id":"612f133a8479a35c282e5ac655f8ad049fe36deb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 17:05:25.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6623d9f1fb604143a4dd82e154f36720\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8593b4cfc314d22a7b551ded102d315 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad11813bc19447d5bfc4e4ce6b3ba480 : SUCCESS in 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"1b5245a1ccf7c7f16f9e62de8d6043f79cae2869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 17:51:00.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f26e0f522d64117bced96a942b0efe4\n\n- grenade https://zuul.opendev.org/t/openstack/build/a672b1cf2cbf4a258c10498add4bac94 : SUCCESS in 1h 01m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/967287a66b5f4a2ba2ae112f26e82b91 : SUCCESS in 59m 18s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4bc19628d6fe49ef8b6c8a45be99ef4a : SUCCESS in 1h 01m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f23e31d594c4058aa8fabb016ef56b0 : SUCCESS in 1h 00m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26437871988e477ba22e1df257db2fb0 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6600924a10ed4237a72e2d7e0d1d170b : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fdfd6b7c7634e14a26891c56fd7ede8 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d56e05121a4ff49d8573706452630b : SUCCESS in 5m 29s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/0445b70c4bec4b3b8e91d60f6f38f02c : FAILURE in 25m 03s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/a1923f35a5a0432e88f176549e50a98d : FAILURE in 17m 41s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/c2e336a1ba6941bf952df1eed01e0d18 : FAILURE in 34m 21s","accounts_in_message":[],"_revision_number":18},{"id":"6d585d8a56004f2678846c36b6c40da5ac8ebc59","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-29 05:49:24.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"6df4d8fd321101b4f058c95a42e80284805b44b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 06:03:37.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55edeb77f0534a9c8ce8faebcd2f087d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23f2cb26ea9946fa9cab5b772722483a : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ea61a1a131e440e9c6c5be84b57a25a : SUCCESS in 11m 52s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"02df1aa873b23da9df629b4473d235845f64688a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 06:55:29.000000000","message":"Patch Set 18:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/868742 is needed.","accounts_in_message":[],"_revision_number":18},{"id":"c0b3e8a735da42797ac7730962890e48bb93a0c2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-29 06:56:00.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"366453784daff25dbb4c3dd66596c3a7fc7d3f1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 07:14:15.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/860dfdbce4884a7abf4b59d8905b704c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8089ce56482542c996719b8cc9c7e533 : SUCCESS in 16m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6a54fdf14c4b4fdea571153e4913569c : SUCCESS in 13m 03s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"deb3dad5fa89c33e90bb4b1ecfe329fc34b304ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 07:40:43.000000000","message":"Patch Set 18:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/868742 is needed.","accounts_in_message":[],"_revision_number":18},{"id":"a86207635962933373641df49ffb26ada14864a6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-29 07:40:45.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"90660f764eea6da9cd650d5054b7441afa7b0399","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 07:53:50.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd7d6785de89420d838d5b9908584b71\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a5293712b6f4d4e8e9d68272cef46a1 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/85a84698b7b2475a875f91c1188ad817 : SUCCESS in 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"704d4cf92c9ce9b72d34dd84549db5b10d3927bf","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-29 08:44:27.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"cb42fad0e1c12d1db69400a22d5210204ec4ac43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-29 08:44:51.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"49ddc66c71c3a24a317935c8132b06a5b94ca6e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 09:00:22.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/228be4cb7bb34170acf1f9b7031eca88\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50e5a480da4d475bb6a39ebf3df706d1 : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d18abfe2c0c4b38a88fe1f022451c86 : SUCCESS in 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"291bea702e0ca669cb24104c0ef68570aa3a0f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 10:03:52.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdb02372a687492dab1aef41a4b39238\n\n- grenade https://zuul.opendev.org/t/openstack/build/bdb93cb744a74cfb8c19a898e256ee85 : SUCCESS in 48m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2d8a1c27e4bf443aaa158722d9ac5a9d : SUCCESS in 46m 26s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/783219d0668841a285465e91147eaf96 : SUCCESS in 1h 17m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3efe6b3aa71248d99e4435592c0bcc7e : SUCCESS in 50m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f6151699754a40827658ab3ba9e523 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d069bdf5743f4d1dbc36d7b0cc6e0e68 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f6cfab400fd4cabb4df453eaf880692 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f648bd0337d24368835123c764826358 : SUCCESS in 6m 06s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/fee9213c340f4942ba6e2f15af59ac40 : FAILURE in 35m 08s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/bf8115e206154c7384b18824efa21a5a : FAILURE in 35m 55s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/1a844eac4a4c47749f05805aca954fd7 : FAILURE in 20m 39s","accounts_in_message":[],"_revision_number":19},{"id":"c35b05027d3b23b9b27774b94d907a1b981deb88","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-29 11:58:05.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e6da698b221e77c8285d325b645e3dbf0d66d87f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 12:10:59.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4babd92f26ea4ac6aaa0498fccbf8b57\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39475dfb68ab4fbea0f0a837203bab7e : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/88062dec6a954561a3f4e49080402a1e : SUCCESS in 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"07cf96c3788c1f89e16332e24ddee668892d6f9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 13:20:45.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8950e939bd204fb6b3444209bfbfaffc\n\n- grenade https://zuul.opendev.org/t/openstack/build/1c4966a3c35f40abbc74ade496f510d2 : SUCCESS in 39m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9eb8eec95ea24f77bae9597f8a30c077 : SUCCESS in 46m 22s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/482e9c6e07754266afb5d2a022b7e1c5 : SUCCESS in 1h 07m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70a1ba60fd7a439099f698bff0f62637 : SUCCESS in 43m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c248919e972419e8db5510f67b9e8de : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f576b5a61124bf7b96b90e5e1d754df : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cb567b4a19b4d95a2fbc994abf559a3 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c69f674a514d9ea94588b342e609f6 : SUCCESS in 5m 24s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/ed836c95f11b44168c659edd380f55f9 : SUCCESS in 1h 21m 03s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/c9d3a22f45ee4a249c3fab3f0245ed11 : FAILURE in 49m 46s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/8d455be9d3e94821bcb023fd13332ce9 : FAILURE in 26m 59s","accounts_in_message":[],"_revision_number":19},{"id":"bf76dc085b7c0e1aff53eac1757fc75d9e1f8566","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-02 14:26:21.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"cdbd7bdf30a6aaae713cfab25a0cd928428104c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 14:27:10.000000000","message":"Patch Set 20:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027experimental\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate-storage\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          # - openstack-tox-functional-py38-fips\n          # - openstack-tox-functional-py39\n          # - glance-tox-functional-py39-rbac-defaults\n          # - glance-code-constants-check\n          # - glance-ceph-thin-provisioning:\n          #     voting: false\n          #     irrelevant-files: \u0026tempest-irrelevant-files\n          #       - ^(test-|)requirements.txt$\n          #       - ^.*\\.rst$\n          #       - ^api-ref/.*$\n          #       - ^glance/hacking/.*$\n          #       - ^glance/locale/.*$\n          #       - ^glance/tests/.*$\n          #       - ^doc/.*$\n          #       - ^releasenotes/.*$\n          #       - ^tools/.*$\n          #       - ^tox.ini$\n          #       - ^\\.zuul\\.yaml$\n          # - tempest-integrated-storage:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-integrated-storage-import:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-integrated-storage-import-standalone:\n          #     irrelevant-files: *tempest-irrelevant-files\n          - glance-multistore-cinder-import-lvm:\n              irrelevant-files: \u0026tempest-irrelevant-files\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^glance/hacking/.*$\n                - ^glance/locale/.*$\n                - ^glance/tests/.*$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^tools/.*$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - glance-multistore-cinder-import-nfs:\n              irrelevant-files: *tempest-irrelevant-files\n          - glance-multistore-cinder-import-rbd:\n              irrelevant-files: *tempest-irrelevant-files\n          # - grenade:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-ipv6-only:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - nova-ceph-multistore:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - glance-secure-rbac-protection-functional\n  \n      gate:\n        jobs:\n          # - openstack-tox-functional-py39\n          # - tempest-integrated-storage:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-integrated-storage-import:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-integrated-storage-import-standalone:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - grenade:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - grenade-skip-level:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-ipv6-only:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - nova-ceph-multistore:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - glance-secure-rbac-protection-functional\n          - glance-multistore-cinder-import-lvm:\n              irrelevant-files: *tempest-irrelevant-files\n          - glance-multistore-cinder-import-nfs:\n              irrelevant-files: *tempest-irrelevant-files\n          - glance-multistore-cinder-import-rbd:\n              irrelevant-files: *tempest-irrelevant-files\n      experimental:\n        jobs:\n          # - glance-tox-py38-glance_store-tips\n          # - glance-tox-functional-py39-glance_store-tips\n          # - glance-tox-functional-py38-glance_store-tips\n          # - barbican-tempest-plugin-simple-crypto\n          # - grenade-multinode\n          # - tempest-pg-full:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - glance-multistore-cinder-import-fips:\n          #     irrelevant-files: *tempest-irrelevant-files\n      periodic:\n        jobs:\n          # NOTE(rosmaita): we only want the \"tips\" jobs to be run against\n          # master, hence the \u0027branches\u0027 qualifiers below.  Without them, when\n          # a stable branch is cut, the tests would be run against the stable\n          # branch as well, which is pointless because these libraries are\n          # frozen (more or less) in the stable branches.\n          #\n          # The \"tips\" jobs can be removed from the stable branch .zuul.yaml\n          # files if someone is so inclined, but that would require manual\n          # maintenance, so we do not do it by default.  Another option is\n          # to define these jobs in the openstack/project-config repo.\n          # That would make us less agile in adjusting these tests, so we\n          # aren\u0027t doing that either.\n          # - glance-tox-functional-py38-oslo-tips:\n          #     branches: master\n          # - glance-tox-py39-keystone-tips:\n          #     branches: master\n          # - glance-tox-py38-keystone-tips:\n          #     branches: master\n          # - glance-tox-functional-py39-keystone-tips:\n          #     branches: master\n          # - glance-tox-functional-py38-keystone-tips:\n          #     branches: master\n          # - glance-tox-py39-glance_store-tips:\n          #     branches: master\n          # - glance-tox-py38-glance_store-tips:\n          #     branches: master\n          # - glance-tox-functional-py39-glance_store-tips:\n          #     branches: master\n          # - glance-tox-functional-py38-glance_store-tips:\n          #     branches: master\n          # - glance-tox-py39-cursive-tips:\n          #     branches: master\n          # - glance-tox-py38-cursive-tips:\n          #     branches: master\n          # - glance-tox-functional-py39-cursive-tips:\n          #     branches: master\n          # - glance-tox-functional-py38-cursive-tips:\n          #     branches: master\n          # - glance-multistore-cinder-import-fips:\n          #     branches: master\n\n  in \"openstack/glance/.zuul.yaml@master\", line 357, column 3","accounts_in_message":[],"_revision_number":20},{"id":"6f18ea056ce8e42df86294e0a72e24eff397f8d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 14:27:16.000000000","message":"Patch Set 20: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027experimental\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate-storage\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          # - openstack-tox-functional-py38-fips\n          # - openstack-tox-functional-py39\n          # - glance-tox-functional-py39-rbac-defaults\n          # - glance-code-constants-check\n          # - glance-ceph-thin-provisioning:\n          #     voting: false\n          #     irrelevant-files: \u0026tempest-irrelevant-files\n          #       - ^(test-|)requirements.txt$\n          #       - ^.*\\.rst$\n          #       - ^api-ref/.*$\n          #       - ^glance/hacking/.*$\n          #       - ^glance/locale/.*$\n          #       - ^glance/tests/.*$\n          #       - ^doc/.*$\n          #       - ^releasenotes/.*$\n          #       - ^tools/.*$\n          #       - ^tox.ini$\n          #       - ^\\.zuul\\.yaml$\n          # - tempest-integrated-storage:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-integrated-storage-import:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-integrated-storage-import-standalone:\n          #     irrelevant-files: *tempest-irrelevant-files\n          - glance-multistore-cinder-import-lvm:\n              irrelevant-files: \u0026tempest-irrelevant-files\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^glance/hacking/.*$\n                - ^glance/locale/.*$\n                - ^glance/tests/.*$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^tools/.*$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - glance-multistore-cinder-import-nfs:\n              irrelevant-files: *tempest-irrelevant-files\n          - glance-multistore-cinder-import-rbd:\n              irrelevant-files: *tempest-irrelevant-files\n          # - grenade:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-ipv6-only:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - nova-ceph-multistore:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - glance-secure-rbac-protection-functional\n  \n      gate:\n        jobs:\n          # - openstack-tox-functional-py39\n          # - tempest-integrated-storage:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-integrated-storage-import:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-integrated-storage-import-standalone:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - grenade:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - grenade-skip-level:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - tempest-ipv6-only:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - nova-ceph-multistore:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - glance-secure-rbac-protection-functional\n          - glance-multistore-cinder-import-lvm:\n              irrelevant-files: *tempest-irrelevant-files\n          - glance-multistore-cinder-import-nfs:\n              irrelevant-files: *tempest-irrelevant-files\n          - glance-multistore-cinder-import-rbd:\n              irrelevant-files: *tempest-irrelevant-files\n      experimental:\n        jobs:\n          # - glance-tox-py38-glance_store-tips\n          # - glance-tox-functional-py39-glance_store-tips\n          # - glance-tox-functional-py38-glance_store-tips\n          # - barbican-tempest-plugin-simple-crypto\n          # - grenade-multinode\n          # - tempest-pg-full:\n          #     irrelevant-files: *tempest-irrelevant-files\n          # - glance-multistore-cinder-import-fips:\n          #     irrelevant-files: *tempest-irrelevant-files\n      periodic:\n        jobs:\n          # NOTE(rosmaita): we only want the \"tips\" jobs to be run against\n          # master, hence the \u0027branches\u0027 qualifiers below.  Without them, when\n          # a stable branch is cut, the tests would be run against the stable\n          # branch as well, which is pointless because these libraries are\n          # frozen (more or less) in the stable branches.\n          #\n          # The \"tips\" jobs can be removed from the stable branch .zuul.yaml\n          # files if someone is so inclined, but that would require manual\n          # maintenance, so we do not do it by default.  Another option is\n          # to define these jobs in the openstack/project-config repo.\n          # That would make us less agile in adjusting these tests, so we\n          # aren\u0027t doing that either.\n          # - glance-tox-functional-py38-oslo-tips:\n          #     branches: master\n          # - glance-tox-py39-keystone-tips:\n          #     branches: master\n          # - glance-tox-py38-keystone-tips:\n          #     branches: master\n          # - glance-tox-functional-py39-keystone-tips:\n          #     branches: master\n          # - glance-tox-functional-py38-keystone-tips:\n          #     branches: master\n          # - glance-tox-py39-glance_store-tips:\n          #     branches: master\n          # - glance-tox-py38-glance_store-tips:\n          #     branches: master\n          # - glance-tox-functional-py39-glance_store-tips:\n          #     branches: master\n          # - glance-tox-functional-py38-glance_store-tips:\n          #     branches: master\n          # - glance-tox-py39-cursive-tips:\n          #     branches: master\n          # - glance-tox-py38-cursive-tips:\n          #     branches: master\n          # - glance-tox-functional-py39-cursive-tips:\n          #     branches: master\n          # - glance-tox-functional-py38-cursive-tips:\n          #     branches: master\n          # - glance-multistore-cinder-import-fips:\n          #     branches: master\n\n  in \"openstack/glance/.zuul.yaml@master\", line 357, column 3","accounts_in_message":[],"_revision_number":20},{"id":"a1f26058b1155c3d956ae58c371f85f1ce14373a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-03 08:23:42.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"7085a4f1e587adb5e6c39ec2b013e06b99fb7305","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 08:37:01.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fcba02fb0d14d2db0dcc2b4fed4dd69\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/071630dcdd5641a2a538c6dfb9681203 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/778c66f891904ba494831d79cc7c7182 : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"fd6e893af46ec9f30839cd7c2bb85e18c096a317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 10:39:59.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a04af342e62e49cdbc2240e4a655c9c7\n\n- grenade https://zuul.opendev.org/t/openstack/build/cdb84b7dc1394a36a4c0751feee9bf70 : SUCCESS in 45m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5dd6154b688e44aea7e9781f10dbf3c7 : SUCCESS in 57m 29s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8bf5594efe16403f9f2e6c9d763104f0 : FAILURE in 1h 05m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5264f27687a6410b8d97aca2301bc4fc : SUCCESS in 52m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8722952bb0148f2bc10982b57504574 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/193f8903d24244ad83dfd19afa12eb55 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4e2f200fc904ef3bfbd7479b1c5cea2 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c55f981d7a041bb8a52949434868dab : SUCCESS in 7m 24s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/57dbc8b408244abda375dbe9d8ef40c2 : TIMED_OUT in 2h 04m 44s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/4b822fdd320441efbae908bb8dd3efd6 : FAILURE in 1h 15m 22s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/12383a8f69434f938417f6ba3dfbed04 : FAILURE in 29m 31s","accounts_in_message":[],"_revision_number":21},{"id":"d205185198c744504f8e77a486d520f4032e88b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-03 12:27:06.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"203b73d4f8d1770df03778ad2da03ddb7691359c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 12:35:03.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcdc292b36424619b6c29602704b5ae7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1f0ad8ba58d4a11981d887b495a9891 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36c00f2440b8440caabe2b0f15452176 : SUCCESS in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"da475beed1a55ea8e514480fb02119fadcfabb04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-03 12:37:09.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"8fb58df3772b124ef9ba2b0cbbfdc2ce9f06bd3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 12:53:03.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9db091e6cc049188da08200826daa09\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e483e91924aa4e4f83f376f9f3ea6f12 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/251af799794144cfa283a16f3fe3fe89 : SUCCESS in 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"bb5cca6f59349e0dcb941a3526d4e652b2e9f32a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-03 14:00:30.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"c41b882faaa8a2eb043885f331a2e00b0a81fa86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 14:18:58.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36d15099ab2f45cc9b7115db654b3e20\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00ba4eb14fbf4bb0bbdac7b2774ac6eb : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cb5cdc49b2764726a960d6787599dc1e : SUCCESS in 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"dce4bd565787a7a175a1f2fa76934fdca6addcc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 15:52:53.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/16ab53772f4745bbaa5d37f388fead21\n\n- grenade https://zuul.opendev.org/t/openstack/build/3430e52b7929451089d86ebb2a38b306 : SUCCESS in 1h 08m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a74f0d8e19114654a3ee520c1a176a20 : SUCCESS in 1h 13m 15s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a151738745548738d7f94f5885d8328 : SUCCESS in 1h 41m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ab2bcb8fa1f49bb8cafc315f9470946 : SUCCESS in 1h 04m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/441bde1190d24469986a356d832bf55c : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60612730105540c3ad296dc4c30ba6a1 : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/375f195b35154cadb9e4e693a89e9fe1 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0cb0002a221497b89a9817f735ff747 : SUCCESS in 7m 56s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/3710226ecd2b40bd88d48b02fb75b7c9 : FAILURE in 1h 13m 05s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/36de9bbcb02f439c8e6da47639c9b048 : FAILURE in 1h 31m 44s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/776548ab5f0a4ee8b6e5bcfb311db98a : FAILURE in 1h 50m 07s","accounts_in_message":[],"_revision_number":24},{"id":"4ede2116f1b4a7d4ee1cb43a8051809518de7d89","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-03 16:12:45.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"b1abed57ca004a5a389c003d97fd4e9c99df0ba7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 16:22:44.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc069f81ef8741a2af93da1c2e415021\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47bca9e5cab241b5b233dd98ef60df2a : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f7d560a6a5449a2b3b8b5fd330bf552 : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"b2dfc29129eb17e4326b6269a7977e4b6506cc71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 17:45:48.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a415f54516146ff9d03ec405278e7c9\n\n- grenade https://zuul.opendev.org/t/openstack/build/8d560436a18e495a9f2587482e662068 : SUCCESS in 57m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/01d3920907b74961aba4fae6169381fc : SUCCESS in 1h 01m 02s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68808ecc14e84a0e8e54111d1583b9c6 : SUCCESS in 1h 30m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/665ee6741ff045f79f29b4ceb7657326 : SUCCESS in 51m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/523d84084af8460d85ffac1f3734e764 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a48c963080d744e8b40f79bc2711e882 : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cce7871a5ce41d099fe34b0af7417ba : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c24ab1e2fb49dfbbc92de4055a1d04 : SUCCESS in 6m 57s\n- glance-multistore-cinder-import-lvm https://zuul.opendev.org/t/openstack/build/b8a198c44ca74d64b663eb61073a161a : SUCCESS in 1h 29m 18s\n- glance-multistore-cinder-import-nfs https://zuul.opendev.org/t/openstack/build/3361b5c11b6741dd8556d9a7642a68cd : FAILURE in 1h 13m 20s\n- glance-multistore-cinder-import-rbd https://zuul.opendev.org/t/openstack/build/d5135ac5789c445eaabb31ddb2b81b5c : FAILURE in 1h 09m 48s","accounts_in_message":[],"_revision_number":24},{"id":"6531d2a94510ff04ddb193c226730e126df5e4d9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-24 07:45:56.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"d2797955173f472f0208933f7af8da2c340cc49f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 07:46:16.000000000","message":"Patch Set 24:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 863977,24","accounts_in_message":[],"_revision_number":24},{"id":"9b17e9796a267d9301a80bd8582861d3366d22a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 07:46:17.000000000","message":"Patch Set 24:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 863977,24","accounts_in_message":[],"_revision_number":24},{"id":"5cd3e5fd85f250bf71deb30406569d9e87db89f7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-24 07:46:17.000000000","message":"Patch Set 24:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/glance for 863977,24","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"04a6ba227fc645269562dde17490453bb39ca80c","revisions":{"779dff6c3216da8eb9b8a54a8484fad9fdd77d83":{"kind":"REWORK","_number":1,"created":"2022-11-08 09:36:12.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/1"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:36:07.000000000","tz":330},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/779dff6c3216da8eb9b8a54a8484fad9fdd77d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/779dff6c3216da8eb9b8a54a8484fad9fdd77d83"}]},"branch":"refs/heads/master"},"2cbc2edad725f13b19f871adedeb45daba9b54a1":{"kind":"REWORK","_number":2,"created":"2022-11-08 11:43:42.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/2"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-11-08 11:43:47.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2cbc2edad725f13b19f871adedeb45daba9b54a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2cbc2edad725f13b19f871adedeb45daba9b54a1"}]},"branch":"refs/heads/master"},"6a7daa3e47b28466763bb16d7088c7cf1fe5a363":{"kind":"REWORK","_number":3,"created":"2022-11-11 08:30:48.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/3"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-11-11 08:30:54.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a7daa3e47b28466763bb16d7088c7cf1fe5a363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a7daa3e47b28466763bb16d7088c7cf1fe5a363"}]},"branch":"refs/heads/master"},"ae7366b41f6b58b2dc562c58ae4f81934731ec1e":{"kind":"REWORK","_number":4,"created":"2022-11-11 08:34:19.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/4"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-11-11 08:34:11.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae7366b41f6b58b2dc562c58ae4f81934731ec1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae7366b41f6b58b2dc562c58ae4f81934731ec1e"}]},"branch":"refs/heads/master"},"69b47a564ade898541a13f181ec09dd5142aec2c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-11-15 08:18:49.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/5"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-11-15 08:18:49.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-nfs/+/763516\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69b47a564ade898541a13f181ec09dd5142aec2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69b47a564ade898541a13f181ec09dd5142aec2c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"99b1b74692cde2c4cf96fa1becf02189e1ace439":{"kind":"REWORK","_number":6,"created":"2022-11-15 08:32:20.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/6"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-11-15 08:32:30.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-nfs/+/763516\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99b1b74692cde2c4cf96fa1becf02189e1ace439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99b1b74692cde2c4cf96fa1becf02189e1ace439"}]},"branch":"refs/heads/master"},"0b8ac21912e4bcbc377e390b29075309c8b7cfe7":{"kind":"REWORK","_number":7,"created":"2022-12-26 08:49:52.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/7"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-26 08:50:31.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-nfs/+/763516\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b8ac21912e4bcbc377e390b29075309c8b7cfe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b8ac21912e4bcbc377e390b29075309c8b7cfe7"}]},"branch":"refs/heads/master"},"8d10c04c0ed1a1f759d4ce6ee8c7b58840c177c8":{"kind":"REWORK","_number":8,"created":"2022-12-26 09:50:01.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/8"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-26 09:50:25.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-nfs/+/763516\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d10c04c0ed1a1f759d4ce6ee8c7b58840c177c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d10c04c0ed1a1f759d4ce6ee8c7b58840c177c8"}]},"branch":"refs/heads/master"},"81c31d3b04411db9a523c68eb626130d35b302e5":{"kind":"REWORK","_number":9,"created":"2022-12-26 10:07:04.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/9"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-26 10:07:29.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-nfs/+/763516\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/81c31d3b04411db9a523c68eb626130d35b302e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/81c31d3b04411db9a523c68eb626130d35b302e5"}]},"branch":"refs/heads/master"},"fc80c05c903a06907c32b79e92d6f7504ca5a1b5":{"kind":"REWORK","_number":10,"created":"2022-12-26 16:28:16.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/10"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-26 16:28:41.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-nfs/+/763516\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc80c05c903a06907c32b79e92d6f7504ca5a1b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc80c05c903a06907c32b79e92d6f7504ca5a1b5"}]},"branch":"refs/heads/master"},"efd1fc96fc0d08318df78b93a84492b926d43562":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2022-12-27 06:18:00.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/11"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-27 06:17:59.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/efd1fc96fc0d08318df78b93a84492b926d43562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/efd1fc96fc0d08318df78b93a84492b926d43562"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c7e7fb8d76b6f4c3c308557fcc909713a3bfd2e9":{"kind":"REWORK","_number":12,"created":"2022-12-27 11:53:57.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/12"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-27 11:53:53.000000000","tz":330},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7e7fb8d76b6f4c3c308557fcc909713a3bfd2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7e7fb8d76b6f4c3c308557fcc909713a3bfd2e9"}]},"branch":"refs/heads/master"},"f08f6fd33eedb92a8b7c804d4345a09d6410e341":{"kind":"REWORK","_number":13,"created":"2022-12-28 04:19:59.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/13"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-28 04:20:24.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f08f6fd33eedb92a8b7c804d4345a09d6410e341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f08f6fd33eedb92a8b7c804d4345a09d6410e341"}]},"branch":"refs/heads/master"},"140a16652431da74e5a0dfe9b679f6d5f794ee3f":{"kind":"REWORK","_number":14,"created":"2022-12-28 07:49:29.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/14"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-28 07:49:52.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/140a16652431da74e5a0dfe9b679f6d5f794ee3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/140a16652431da74e5a0dfe9b679f6d5f794ee3f"}]},"branch":"refs/heads/master"},"ba7bd9253a79ca3ea05339510b06245d17931e60":{"kind":"REWORK","_number":15,"created":"2022-12-28 08:39:23.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/15"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-28 08:39:23.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba7bd9253a79ca3ea05339510b06245d17931e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba7bd9253a79ca3ea05339510b06245d17931e60"}]},"branch":"refs/heads/master"},"1681d7a702bf033e339b5450517c5c9f857b8409":{"kind":"REWORK","_number":16,"created":"2022-12-28 10:46:52.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/16"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-28 10:46:52.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1681d7a702bf033e339b5450517c5c9f857b8409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1681d7a702bf033e339b5450517c5c9f857b8409"}]},"branch":"refs/heads/master"},"9156821d3879406fd605a0badf0cdcabee67f870":{"kind":"REWORK","_number":17,"created":"2022-12-28 15:07:11.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/17"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-28 15:07:11.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/843103\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9156821d3879406fd605a0badf0cdcabee67f870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9156821d3879406fd605a0badf0cdcabee67f870"}]},"branch":"refs/heads/master"},"6c24ff8d79abdddd0d9d8c4a11ab7e12b7c8d0b3":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2022-12-28 16:44:00.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/18"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-28 16:43:50.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/868742\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c24ff8d79abdddd0d9d8c4a11ab7e12b7c8d0b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c24ff8d79abdddd0d9d8c4a11ab7e12b7c8d0b3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"54b3023ed1746ab57d94f5c090038164989930f1":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-12-29 08:44:51.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/19"}}},"commit":{"parents":[{"commit":"e2cd7251f46e9d8fe7796ba0e178ee692de98fe6","subject":"Fix tox4 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2cd7251f46e9d8fe7796ba0e178ee692de98fe6"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-12-29 08:44:51.000000000","tz":0},"subject":"DNM: Test cinder store refactoring","message":"DNM: Test cinder store refactoring\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/868742\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54b3023ed1746ab57d94f5c090038164989930f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54b3023ed1746ab57d94f5c090038164989930f1"}]},"branch":"refs/heads/master","description":"Rebase"},"1af516dcd6968ce201af76f53200c2521ad6e9be":{"kind":"REWORK","_number":20,"created":"2023-02-02 14:26:21.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/20"}}},"commit":{"parents":[{"commit":"907c56265438da43c13bf1a3369d5459b1267e34","subject":"Merge \"Enforce image safety during image_conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/907c56265438da43c13bf1a3369d5459b1267e34"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-02-02 14:26:27.000000000","tz":0},"subject":"DNM: Test cinder store extend attached volumes","message":"DNM: Test cinder store extend attached volumes\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/868742\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1af516dcd6968ce201af76f53200c2521ad6e9be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1af516dcd6968ce201af76f53200c2521ad6e9be"}]},"branch":"refs/heads/master"},"dfa3f12c663ff7975df49db07546dc058379598f":{"kind":"REWORK","_number":21,"created":"2023-02-03 08:23:42.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/21"}}},"commit":{"parents":[{"commit":"907c56265438da43c13bf1a3369d5459b1267e34","subject":"Merge \"Enforce image safety during image_conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/907c56265438da43c13bf1a3369d5459b1267e34"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-02-03 08:24:42.000000000","tz":0},"subject":"DNM: Test cinder store extend attached volumes","message":"DNM: Test cinder store extend attached volumes\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/868742\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfa3f12c663ff7975df49db07546dc058379598f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfa3f12c663ff7975df49db07546dc058379598f"}]},"branch":"refs/heads/master"},"9e3dbe2cea9d474e7895b3192cb027fe1fe93ac6":{"kind":"REWORK","_number":22,"created":"2023-02-03 12:27:06.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/22"}}},"commit":{"parents":[{"commit":"907c56265438da43c13bf1a3369d5459b1267e34","subject":"Merge \"Enforce image safety during image_conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/907c56265438da43c13bf1a3369d5459b1267e34"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-02-03 12:27:51.000000000","tz":0},"subject":"DNM: Test cinder store extend attached volumes","message":"DNM: Test cinder store extend attached volumes\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/868742\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e3dbe2cea9d474e7895b3192cb027fe1fe93ac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e3dbe2cea9d474e7895b3192cb027fe1fe93ac6"}]},"branch":"refs/heads/master"},"4f28e0d4da7fcda3edc227a4021cf36482997f12":{"kind":"REWORK","_number":23,"created":"2023-02-03 12:37:09.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/23"}}},"commit":{"parents":[{"commit":"907c56265438da43c13bf1a3369d5459b1267e34","subject":"Merge \"Enforce image safety during image_conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/907c56265438da43c13bf1a3369d5459b1267e34"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-02-03 12:37:54.000000000","tz":0},"subject":"DNM: Test cinder store extend attached volumes","message":"DNM: Test cinder store extend attached volumes\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/868742\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f28e0d4da7fcda3edc227a4021cf36482997f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f28e0d4da7fcda3edc227a4021cf36482997f12"}]},"branch":"refs/heads/master"},"04a6ba227fc645269562dde17490453bb39ca80c":{"kind":"REWORK","_number":24,"created":"2023-02-03 14:00:30.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/77/863977/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/863977/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/863977/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/863977/24"}}},"commit":{"parents":[{"commit":"907c56265438da43c13bf1a3369d5459b1267e34","subject":"Merge \"Enforce image safety during image_conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/907c56265438da43c13bf1a3369d5459b1267e34"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-08 09:35:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-02-03 14:01:15.000000000","tz":0},"subject":"DNM: Test cinder store extend attached volumes","message":"DNM: Test cinder store extend attached volumes\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/868742\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/868672\n\nChange-Id: Ib7025845656664bbbc44b4b7a160e6dfba4e783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04a6ba227fc645269562dde17490453bb39ca80c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04a6ba227fc645269562dde17490453bb39ca80c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
