)]}'
{"id":"openstack%2Fglance~864128","triplet_id":"openstack%2Fglance~master~I0352ab6b2c00055de094ac5902b8d50941d06dcf","project":"openstack/glance","branch":"master","topic":"bug/1983279","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2022-11-15 04:59:41.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-11-15 04:59:41.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-11-15 03:06:28.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-11-15 04:59:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0352ab6b2c00055de094ac5902b8d50941d06dcf","subject":"Fix unintentional exception inspecting VMDK","status":"MERGED","created":"2022-11-09 15:49:02.000000000","updated":"2022-11-15 05:00:56.000000000","submitted":"2022-11-15 04:59:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864128-bug/1983279","meta_rev_id":"86b5b38e90af69108f3b18330a97b22d5129c896","_number":864128,"virtual_id_number":864128,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-15 04:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-11-15 03:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-11-14 16:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-11-15 03:06:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-09 17:27:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-14 14:39:53.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-11-14 14:39:53.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2022-11-14 14:39:53.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-11-14 16:44:22.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"0cef7a0197c2cb445a951e62dd4111361488d45b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-09 15:49:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e7534354774c2c2e091467f02ecac69ba856637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 17:27:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e129a369d874c55af7de16218f9689e\n\n- grenade https://zuul.opendev.org/t/openstack/build/9883309809834537b962022c2c3fde18 : SUCCESS in 43m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f0950890df8a4dd8be22dbce6d8ca420 : SUCCESS in 47m 17s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ea249dd825f49aca7d85c85c7f38e1f : SUCCESS in 1h 22m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e06242d9fffc45e5adaf5051e93dec8a : SUCCESS in 1h 07m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f00078e267b4a4aa779ae91d640d7c5 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0932d995e3f48258b0151c714b17ad6 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3c230514ebb4525901a1aef12a3c40b : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc185b9227b4ce6ab434caa0f9ac1cb : SUCCESS in 4m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7b7be1316b0f4cfb8165c0c4d367a0f6 : SUCCESS in 10m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2bd8937ce1894f8f8f8a60b205197629 : SUCCESS in 8m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/57e7f8a58702429a85e8da1392ccd420 : SUCCESS in 9m 14s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5f32b4a721444a3cb53f3e803e26ed10 : SUCCESS in 4m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9c5d879e9b804125b5cd011121f24406 : SUCCESS in 1h 15m 56s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/423a406ed01f453eb09fa4981c5a32f1 : SUCCESS in 1h 16m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/471c09ee28e548e68e64a6f013f349e1 : SUCCESS in 1h 16m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c252fd2c994d499f91ede9f60bfc5050 : SUCCESS in 1h 19m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/327b57201841421a8fa079deec287d60 : SUCCESS in 57m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0eac41b5ca65408995947ab51dc92aa9 : SUCCESS in 1h 07m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/83fe1d067c1840bcbd9db9f96752a19b : SUCCESS in 28m 26s","accounts_in_message":[],"_revision_number":1},{"id":"66e79d9d2a57468001b07ee334bb6643be9ab7cb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-14 14:39:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5000b0734ff6019c723084e35acecffef9b6230c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-11-14 16:44:22.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c856cef3d415d4bfef99730a8d036e62b3c7a28","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-11-15 03:06:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfb1fba1147255cc0ab56c9ca0ebf13ee2ae6752","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 03:07:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71d71ac063bbec4f598bfcad454ee31b0026bc9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 04:59:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/836f742df9854cdfaa89d5c32bd49d0f\n\n- grenade https://zuul.opendev.org/t/openstack/build/a58950de4e2e42b7b256ce37ee1db2dc : SUCCESS in 53m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5eb17f0449ed4eb982eac603a886be89 : SUCCESS in 1h 47m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b22383f8a794a06b9df3814bd450b07 : SUCCESS in 54m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77525ea77edd4779ad1abbed903b3fe5 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c6ee80f0cdf4254b28b271a003e4c53 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a55c92ea04084a39bcdb742965de5fd5 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2823e5961e16451aad0cb5cedf6797f2 : SUCCESS in 6m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6e42f58164fe478a99a9e86cdde44d08 : SUCCESS in 7m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d42c31ea7f1344e4a204f391d3ee0caf : SUCCESS in 9m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/31bb4d68bb81407f905c224576a81e70 : SUCCESS in 1h 20m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/52593a6facbd423298a684e3c474d3a9 : SUCCESS in 1h 12m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b62e59466cf64888ae2d25d7a13baad3 : SUCCESS in 51m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d58619b00205437594e24819e14bb3ca : SUCCESS in 50m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dddab3ff53e446c9802a8688b012d948 : SUCCESS in 58m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fc56a202751c4e9c86154ac37b5483f3 : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":1},{"id":"610862513b87ad3ee0d6e0e65386d14eb2f442d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 04:59:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"86b5b38e90af69108f3b18330a97b22d5129c896","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 05:00:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b265f16dfb674f2eaf6e572b977a7d49\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d961d57766b443dbce8b5f56b4018f7 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"199722a65a88f9c28b554b180ec36415ed3a51ac","revisions":{"199722a65a88f9c28b554b180ec36415ed3a51ac":{"kind":"REWORK","_number":1,"created":"2022-11-09 15:49:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/28/864128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/864128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/864128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/864128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/864128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/864128/1"}}},"commit":{"parents":[{"commit":"3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a","subject":"Merge \"Fix a document description error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b37f17d3bdec0038fdca5d639e24ede0ffa6c0a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-09 15:33:37.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-09 15:35:54.000000000","tz":-480},"subject":"Fix unintentional exception inspecting VMDK","message":"Fix unintentional exception inspecting VMDK\n\nIt looks like a raise statement was left in the virtual_size property\nhandler for VMDK, which should have been converted to a log at some\npoint. All the other inspectors return zero for virtual_size if the\nformat does not match or they are unable to parse the data. This\nconverts that raise to a log, and adds a test to make sure we make it\nfar enough in the processing of the complex VMDK format to ensure that\nbehavior.\n\nCloses-Bug: #1983279\n\nChange-Id: I0352ab6b2c00055de094ac5902b8d50941d06dcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/199722a65a88f9c28b554b180ec36415ed3a51ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/199722a65a88f9c28b554b180ec36415ed3a51ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
