)]}'
{"id":"openstack%2Fglance~864865","triplet_id":"openstack%2Fglance~stable%2Fwallaby~Ic4c3168bbbce8f37d4212b4f131fe6d959afd519","project":"openstack/glance","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2023-02-09 13:50:33.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-11-18 09:29:18.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"Ic4c3168bbbce8f37d4212b4f131fe6d959afd519","subject":"[DNM] Test only SRBAC job with pinned glance tempest","status":"ABANDONED","created":"2022-11-17 10:27:36.000000000","updated":"2023-02-09 13:50:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ec3259b1e403975b3eda7e2066b9f3e09022b647","_number":864865,"virtual_id_number":864865,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-18 11:19:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-17 10:28:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"235d436db2349789c911c274dd3506df7ff534eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-11-17 10:27:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"550b85aaf86c4bc11fdae76dcc542a9bc0006c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 10:28:13.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch stable/wallaby.  The error was:\n\n  found undefined alias\n  in \"openstack/glance/.zuul.yaml@stable/wallaby\", line 306, column 31","accounts_in_message":[],"_revision_number":1},{"id":"374035e73202c1556cc5a0a3ad2a0c70e81c68bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-11-17 11:16:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"15dc6e3461fb7cbff8d061e6c4b295a5156f13d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-11-17 11:52:48.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"24ea34ebed53f3b6cf8d7737c3234ac5701bf655","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-11-17 12:39:46.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"73f9baae7a808f6d6b369e85016f46593f583861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:06:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7322effc16944dc3b93a4efe5f9ab9da\n\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1721073d1f3946229b2d98584cdd1542 : FAILURE in 24m 20s","accounts_in_message":[],"_revision_number":4},{"id":"4b9818003ec922d91b64aecc102276ee290d8a2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-11-18 05:36:12.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"289ae4f0f46be354feee213db51df89717dfacd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 06:06:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/032ee4a17ed94f60972e6825490571b7\n\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1d83a5a075ef4df4bfde3168c4f72832 : FAILURE in 27m 36s","accounts_in_message":[],"_revision_number":5},{"id":"96fe8f73522576b4dbbd9fbb8104cd943cdc2e54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-11-18 08:42:09.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6274f5570abc03987e905b95cea39a6b1884425f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 09:29:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6482eedfb794bc389c7c0770a7a28aa\n\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6d479e15f34b49bdabba186e3de26dfa : FAILURE in 41m 45s","accounts_in_message":[],"_revision_number":6},{"id":"2d06e340719f6a3e418bf0677e6a60f470ee6474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-11-18 10:19:37.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"a5feebb91a5bffc5ecfc4c8e5d8c8e16096b98a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 10:45:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a327fa3e4e34e08ba9466e255f429d0\n\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a40f55129c8c4c76b8d31a2915eb8293 : FAILURE in 24m 28s","accounts_in_message":[],"_revision_number":7},{"id":"40eb6a26f6b7607b9b3cb09f674f5483b52646a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-11-18 10:51:24.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"302b88c90f453631d2e8002c2e289a7260470e0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 11:19:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bdea6cd734545b18212201240ccc920\n\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ea9e4e33849d4681b7f06042cd8c7a0c : SUCCESS in 25m 22s","accounts_in_message":[],"_revision_number":8},{"id":"ec3259b1e403975b3eda7e2066b9f3e09022b647","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-09 13:50:33.000000000","message":"Abandoned\n\nNot needed any more","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"5b57bbde9f3c0a8367b309c735a1bff744f2d6fc","revisions":{"90db6381088e29ffc0317687b684d9d68d9b4a3f":{"kind":"REWORK","_number":1,"created":"2022-11-17 10:27:36.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/65/864865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/864865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/864865/1"}}},"commit":{"parents":[{"commit":"677c89c23631e9083261a1a18ed438d8966e0de2","subject":"Fix set_property_atomic() boolean type casting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/677c89c23631e9083261a1a18ed438d8966e0de2"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2022-11-17 10:03:11.000000000","tz":330},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2022-11-17 10:25:36.000000000","tz":330},"subject":"[DNM] Test only SRBAC job","message":"[DNM] Test only SRBAC job\n\nDepends-On: I232bc47b77314f6fc4cb192530897cadd75a5426\nChange-Id: Ic4c3168bbbce8f37d4212b4f131fe6d959afd519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90db6381088e29ffc0317687b684d9d68d9b4a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90db6381088e29ffc0317687b684d9d68d9b4a3f"}]},"branch":"refs/heads/stable/wallaby"},"71910c3f78df4a5b6be6dd9543d8f9ef728fe3f3":{"kind":"REWORK","_number":2,"created":"2022-11-17 11:16:58.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/65/864865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/864865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/864865/2"}}},"commit":{"parents":[{"commit":"677c89c23631e9083261a1a18ed438d8966e0de2","subject":"Fix set_property_atomic() boolean type casting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/677c89c23631e9083261a1a18ed438d8966e0de2"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2022-11-17 10:03:11.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-11-17 11:16:57.000000000","tz":0},"subject":"[DNM] Test only SRBAC job","message":"[DNM] Test only SRBAC job\n\nDepends-On: I232bc47b77314f6fc4cb192530897cadd75a5426\nChange-Id: Ic4c3168bbbce8f37d4212b4f131fe6d959afd519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71910c3f78df4a5b6be6dd9543d8f9ef728fe3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71910c3f78df4a5b6be6dd9543d8f9ef728fe3f3"}]},"branch":"refs/heads/stable/wallaby"},"8219ca54bbaf866b44519eb7e114e41ea4052368":{"kind":"REWORK","_number":3,"created":"2022-11-17 11:52:48.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/65/864865/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/864865/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/864865/3"}}},"commit":{"parents":[{"commit":"677c89c23631e9083261a1a18ed438d8966e0de2","subject":"Fix set_property_atomic() boolean type casting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/677c89c23631e9083261a1a18ed438d8966e0de2"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2022-11-17 10:03:11.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-11-17 11:52:47.000000000","tz":0},"subject":"[DNM] Test only SRBAC job","message":"[DNM] Test only SRBAC job\n\nDepends-On: I232bc47b77314f6fc4cb192530897cadd75a5426\nChange-Id: Ic4c3168bbbce8f37d4212b4f131fe6d959afd519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8219ca54bbaf866b44519eb7e114e41ea4052368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8219ca54bbaf866b44519eb7e114e41ea4052368"}]},"branch":"refs/heads/stable/wallaby"},"30e4d7230b416bb591e2f81e25ae19d4a10a85d0":{"kind":"REWORK","_number":4,"created":"2022-11-17 12:39:46.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/65/864865/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/864865/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/864865/4"}}},"commit":{"parents":[{"commit":"677c89c23631e9083261a1a18ed438d8966e0de2","subject":"Fix set_property_atomic() boolean type casting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/677c89c23631e9083261a1a18ed438d8966e0de2"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2022-11-17 10:03:11.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-11-17 12:39:46.000000000","tz":0},"subject":"[DNM] Test only SRBAC job","message":"[DNM] Test only SRBAC job\n\nDepends-On: I20cb72e1b44a261e43839ed3056b4d2ae287c2ff\nChange-Id: Ic4c3168bbbce8f37d4212b4f131fe6d959afd519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30e4d7230b416bb591e2f81e25ae19d4a10a85d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30e4d7230b416bb591e2f81e25ae19d4a10a85d0"}]},"branch":"refs/heads/stable/wallaby"},"851c7b9c532fbe97b82bdc2d2846bbed17bfbb49":{"kind":"REWORK","_number":5,"created":"2022-11-18 05:36:12.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/65/864865/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/864865/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/864865/5"}}},"commit":{"parents":[{"commit":"677c89c23631e9083261a1a18ed438d8966e0de2","subject":"Fix set_property_atomic() boolean type casting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/677c89c23631e9083261a1a18ed438d8966e0de2"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2022-11-17 10:03:11.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-11-18 05:36:12.000000000","tz":0},"subject":"[DNM] Test only SRBAC job with pinned glance tempest","message":"[DNM] Test only SRBAC job with pinned glance tempest\n\nChange-Id: Ic4c3168bbbce8f37d4212b4f131fe6d959afd519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/851c7b9c532fbe97b82bdc2d2846bbed17bfbb49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/851c7b9c532fbe97b82bdc2d2846bbed17bfbb49"}]},"branch":"refs/heads/stable/wallaby"},"e20ca8667c1cc9d409050964229750efd38bd235":{"kind":"REWORK","_number":6,"created":"2022-11-18 08:42:09.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/65/864865/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/864865/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/864865/6"}}},"commit":{"parents":[{"commit":"677c89c23631e9083261a1a18ed438d8966e0de2","subject":"Fix set_property_atomic() boolean type casting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/677c89c23631e9083261a1a18ed438d8966e0de2"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2022-11-17 10:03:11.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-11-18 08:42:08.000000000","tz":0},"subject":"[DNM] Test only SRBAC job with pinned glance tempest","message":"[DNM] Test only SRBAC job with pinned glance tempest\n\nDepends-On: I20cb72e1b44a261e43839ed3056b4d2ae287c2ff\nChange-Id: Ic4c3168bbbce8f37d4212b4f131fe6d959afd519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e20ca8667c1cc9d409050964229750efd38bd235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e20ca8667c1cc9d409050964229750efd38bd235"}]},"branch":"refs/heads/stable/wallaby"},"0918005879897dff996d9933443e031d0e03e45f":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-11-18 10:19:37.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/65/864865/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/864865/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/864865/7"}}},"commit":{"parents":[{"commit":"677c89c23631e9083261a1a18ed438d8966e0de2","subject":"Fix set_property_atomic() boolean type casting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/677c89c23631e9083261a1a18ed438d8966e0de2"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2022-11-17 10:03:11.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-11-18 10:19:37.000000000","tz":0},"subject":"[DNM] Test only SRBAC job with pinned glance tempest","message":"[DNM] Test only SRBAC job with pinned glance tempest\n\nDepends-On: I09f513d08212bc80a3a86a750b29b1c6625d2f89\nChange-Id: Ic4c3168bbbce8f37d4212b4f131fe6d959afd519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0918005879897dff996d9933443e031d0e03e45f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0918005879897dff996d9933443e031d0e03e45f"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"5b57bbde9f3c0a8367b309c735a1bff744f2d6fc":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-11-18 10:51:24.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/65/864865/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/864865/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/864865/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/864865/8"}}},"commit":{"parents":[{"commit":"677c89c23631e9083261a1a18ed438d8966e0de2","subject":"Fix set_property_atomic() boolean type casting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/677c89c23631e9083261a1a18ed438d8966e0de2"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2022-11-17 10:03:11.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-11-18 10:51:24.000000000","tz":0},"subject":"[DNM] Test only SRBAC job with pinned glance tempest","message":"[DNM] Test only SRBAC job with pinned glance tempest\n\nDepends-On: If9321ec7707b988c01a77e1a0ec67f62b53eedc8\nChange-Id: Ic4c3168bbbce8f37d4212b4f131fe6d959afd519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b57bbde9f3c0a8367b309c735a1bff744f2d6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b57bbde9f3c0a8367b309c735a1bff744f2d6fc"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
