)]}'
{"id":"openstack%2Fglance~864910","triplet_id":"openstack%2Fglance~stable%2Fxena~I0352ab6b2c00055de094ac5902b8d50941d06dcf","project":"openstack/glance","branch":"stable/xena","topic":"xena/bug/1983279","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-12-15 16:16:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0352ab6b2c00055de094ac5902b8d50941d06dcf","subject":"Fix unintentional exception inspecting VMDK","status":"MERGED","created":"2022-11-17 15:10:53.000000000","updated":"2022-12-15 16:18:03.000000000","submitted":"2022-12-15 16:16:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864910-xena/bug/1983279","meta_rev_id":"c62b6f8c607ea5be10a2b1eae383ed9ba611a189","_number":864910,"virtual_id_number":864910,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-15 16:16:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-12-15 14:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-11-21 20:44:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-12-15 14:38:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-17 16:45:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-21 20:44:06.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-12-15 14:38:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"6185f8e707b4a8358c0d617fe8b13a241b290fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-17 15:10:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71cb12546c52e0b3edf39ef3ed51b03f7de2bea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 16:45:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39852b0bfadf478c8d8ce1b66d1cf74b\n\n- grenade https://zuul.opendev.org/t/openstack/build/cc21f349983442708b6f3a4d7c9fa0c7 : SUCCESS in 1h 04m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c93e806088a44cb81e3eddd053f2e24 : SUCCESS in 1h 18m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9db29c1667094f519a6521f8510fef36 : SUCCESS in 47m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5eb016d405940ca8ef2ace65ea5b69f : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04732734c7fc4a89b0f51f36ec7fccb4 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99ebe4127c274a8ca94e2b442f1deb05 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/914fe60f0d034c7e8da53a20745079cf : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a6152b714924e5781036d23b1f6988b : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6ca257608b174fd38dce48f81073ff3a : SUCCESS in 9m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/733c2fcebb1c49c5a297d8037322982f : SUCCESS in 7m 43s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3341d6b896b3460b8224e8cba80019a4 : SUCCESS in 8m 39s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3d06eb22f91543128c35631c25824476 : SUCCESS in 1h 00m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/54671b80e644461da70171a4ff80960a : SUCCESS in 1h 09m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aff187d214494426b4b823b8d16568ef : SUCCESS in 1h 28m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5c5a9e8833e64976ae3f22cf4965c48a : SUCCESS in 1h 13m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3ce6489b4464da0ab5c3ea6a996464d : SUCCESS in 43m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/961af0fb98724f3c8d9395477b9f5653 : SUCCESS in 1h 04m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b2543a23e8404eaaab678c545d663b6a : SUCCESS in 29m 45s","accounts_in_message":[],"_revision_number":1},{"id":"335f5e844b9bda3ce2ba1ad55c019e3e26aca83c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-11-21 20:44:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"045c0123eb5dcd368dcb82d9ed66c071addae6ab","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-12-15 14:38:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"500a1e07b88205d175f9a9584b025286f16ad35e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 14:39:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc2682716f8b7f278dbaf32cca57caede6602366","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 16:16:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/543960f8118d41669b36a3bd43b2a465\n\n- grenade https://zuul.opendev.org/t/openstack/build/6f51796442bc46e7b7104653302c0a6f : SUCCESS in 58m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1de11f65f8db4c88a469eaf22cb1d133 : SUCCESS in 1h 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c889da0aace745dfaa37b8f8693ed493 : SUCCESS in 47m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28abba8678404028b990fc640e923521 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12c3af8af3294869991aae69919a08d2 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83163fd42eae449d935f23300b9082de : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b8d1762bb0445cc9aa104be586b5a92 : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1729e77322a34b179abeb8aefd3b8ce0 : SUCCESS in 9m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89133c6a59634667bca1fed0c4f7d332 : SUCCESS in 8m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4dd9e1847eca43c68dd7e213929febc8 : SUCCESS in 1h 11m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/11318e05003d45189ecb1c5d0fec41cd : SUCCESS in 1h 32m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d57385380f004caf99b17723cc3f2092 : SUCCESS in 49m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1fe07bc277814140a1a00cd1caeea6b9 : SUCCESS in 1h 19m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/98a5ef87b2254f19ac46a4f43d413245 : SUCCESS in 26m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2fb1e61d15eba346d29b9e1df20bc83467feb9c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 16:16:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c62b6f8c607ea5be10a2b1eae383ed9ba611a189","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 16:18:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeb28ded90d9484e807caa497d6614f3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61ef6557e21442c8b4ed486c0c070c60 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2281c0868e00d0289e4a5c78fbff99f67b5d3d4d","revisions":{"2281c0868e00d0289e4a5c78fbff99f67b5d3d4d":{"kind":"REWORK","_number":1,"created":"2022-11-17 15:10:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/10/864910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/864910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/864910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/864910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/864910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/864910/1"}}},"commit":{"parents":[{"commit":"3cfbb69c017ed0e2c61f267829a3b79b073832f0","subject":"[stable-only] Remove glance-code-constants-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3cfbb69c017ed0e2c61f267829a3b79b073832f0"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-09 15:33:37.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-17 15:10:44.000000000","tz":-480},"subject":"Fix unintentional exception inspecting VMDK","message":"Fix unintentional exception inspecting VMDK\n\nIt looks like a raise statement was left in the virtual_size property\nhandler for VMDK, which should have been converted to a log at some\npoint. All the other inspectors return zero for virtual_size if the\nformat does not match or they are unable to parse the data. This\nconverts that raise to a log, and adds a test to make sure we make it\nfar enough in the processing of the complex VMDK format to ensure that\nbehavior.\n\nCloses-Bug: #1983279\n\nChange-Id: I0352ab6b2c00055de094ac5902b8d50941d06dcf\n(cherry picked from commit 199722a65a88f9c28b554b180ec36415ed3a51ac)\n(cherry picked from commit 07f98747a2d4bf65f44348603ecba639d602db4d)\n(cherry picked from commit 323e9f5f93c99121c6ee34e711fb4fd1d191e430)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2281c0868e00d0289e4a5c78fbff99f67b5d3d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2281c0868e00d0289e4a5c78fbff99f67b5d3d4d"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
