)]}'
{"id":"openstack%2Fglance~864912","triplet_id":"openstack%2Fglance~stable%2Fwallaby~I0352ab6b2c00055de094ac5902b8d50941d06dcf","project":"openstack/glance","branch":"stable/wallaby","topic":"wallaby/bug/1983279","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-02-09 14:58:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0352ab6b2c00055de094ac5902b8d50941d06dcf","subject":"Fix unintentional exception inspecting VMDK","status":"ABANDONED","created":"2022-11-17 15:11:34.000000000","updated":"2023-02-09 14:58:06.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2eed2f24502ec4f7e829d61c021264b7c61638c1","_number":864912,"virtual_id_number":864912,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-17 16:31:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-17 16:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-21 20:51:38.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2023-02-06 19:11:56.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"}],"messages":[{"id":"4e7389baf11d54f60b3e410693c01d1899a8f939","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-17 15:11:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25f608322d1809a5aea7c60ece616ff5a76b67b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 16:31:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba2dabb17b35400e91347b0a1420febc\n\n- grenade https://zuul.opendev.org/t/openstack/build/cbbaba92a73e42729e0d7f9eaccea4b7 : SUCCESS in 1h 01m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/95ba12e693b747a4b99a66566019c82e : SUCCESS in 1h 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53d6481decb04905a2f8faae85e97a30 : SUCCESS in 51m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d94fbb74e35426bb6e55b7cc09e8436 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53086918f83640d7818caaa73ab91a35 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6447e0280dd24baca5ca389648dd9320 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17839e0f602247a9a678d15f06ce1aef : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32e307a83ecf4bd7872ef5c0cf5e1590 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569ac7bc50af45af9238516ec5973efe : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/13219b63231d4c00bf8514fe5892c50c : TIMED_OUT in 31m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/940f5a42bc21469391dd6a7bfb30f426 : SUCCESS in 7m 23s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3b25f11a16db40c9ba0cbc7ab16e3b9e : FAILURE in 3m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9b834201f50d436c96fd8c9fd9902ce0 : FAILURE in 1h 06m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97b67c56dad443b588f0ee9db9ae8b62 : SUCCESS in 1h 14m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b991ed65c4cc42d9ba30ab2b30182c5e : SUCCESS in 1h 12m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8256a0bce4be4885a5fce1897b5f5a98 : SUCCESS in 1h 18m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4402cb1d9ca4405fadbb4953ec4c4d86 : SUCCESS in 56m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d64674cd3a9344119ca617ead9aa0690 : FAILURE in 1h 01m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7082b2a1f88a47a78cbade9b579810f9 : FAILURE in 29m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6afc380f6f9dcfe6a6cf346f2165e6c2c935e401","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-11-21 20:51:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d376a7dc90ceae076b529668162bddbc1d3f49cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 22:15:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e197c1f19c7f4cfba8cbe6c4dc211e06\n\n- grenade https://zuul.opendev.org/t/openstack/build/235f3cd45e734cc991e2a09863c6a20f : SUCCESS in 44m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/34c0e5c1a66a4771a6eeedd90a86ccbc : SUCCESS in 1h 01m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/355dc543f52d4fbca263cdd8a3b9b7b3 : SUCCESS in 35m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58cb8c5589ff4a328278540c616666ec : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d456718622c47b5a3fa89e04ad6c34b : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/134e920abe3846289cb7813595b00691 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c28256408d3244fcb8db64a3ae6f70e4 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/101f0316d8634e9da7c0f573af139f1c : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fccebd235da437ab26de23910930f74 : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db4db4a988f44a0a91cb2c3c06ccc12c : SUCCESS in 8m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a41baf25d094bec8503ccc2b47b95cc : SUCCESS in 8m 36s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/44ad2fe21d5f4a1f852b74a4813eaf46 : FAILURE in 3m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6e65cd0ac0f642ca99ab536f840ae72b : FAILURE in 1h 08m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3be8dd5da7724535bd063085303124da : SUCCESS in 1h 08m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4647ede402b54b8f9d9629d2de74c3a6 : SUCCESS in 1h 16m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/956cb6f5ad3946bc80ef77759d507184 : SUCCESS in 1h 21m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eea4eb19a2c445bab49e76a54ddb712b : SUCCESS in 52m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fe0990a4c81a45e39aea95b2c2129983 : FAILURE in 1h 06m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ab30f5899ed94b2bbc40fba836f06be6 : FAILURE in 27m 03s","accounts_in_message":[],"_revision_number":1},{"id":"91cbc61df8d973f8795e239df1bb28bb5cf5b42c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-06 19:11:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20f17729d389171158e2578dd02e61d724c8f110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 20:52:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/597312fd24fd4b2c9448eda1c2804565\n\n- grenade https://zuul.opendev.org/t/openstack/build/fb366f657e3d471c8a0e96746b041961 : SUCCESS in 1h 07m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/208612c583d2485d8f826516aa363871 : SUCCESS in 52m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/714ff54afa2b44deb39ea5d39bb74ee1 : SUCCESS in 45m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66ce6632cf204ce5a60e3639085a531b : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51127233a989458fa9a449b995a78ff7 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82666da67db34d3c884f2b1c1d05d390 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5201a3a88580437ba1ddaf936a45526c : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/821d752b1a0b4c069d21122496292b67 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dbf1c95566a4f8881b2b3e7575e9d2d : FAILURE in 3m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72ea40b20b2a41d8b5ffc77174076498 : SUCCESS in 9m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e8de915083f540c1be54b230f3d9eb6c : SUCCESS in 8m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f09933b3a9f34359ad56c4e187ece970 : FAILURE in 1h 33m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5a00908b51194177ad0e66c2471c9319 : SUCCESS in 1h 25m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7ace6a22e6e440acb7e08fc0d2f6184c : SUCCESS in 1h 12m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/aae555db64af48ccb20203945bfc9284 : SUCCESS in 56m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88612394a6af413fac52addbb4e3e11b : SUCCESS in 1h 01m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e35b4f1673f5447cba3bb862845acd84 : SUCCESS in 1h 14m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7a9a5b7b299345f4b45957dae94a6da5 : SUCCESS in 23m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2eed2f24502ec4f7e829d61c021264b7c61638c1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-09 14:58:06.000000000","message":"Abandoned\n\nThis branch is too old, we will not merge this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8845a101a30ce010215f0f25f1419d992db7956d","revisions":{"8845a101a30ce010215f0f25f1419d992db7956d":{"kind":"REWORK","_number":1,"created":"2022-11-17 15:11:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/864912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/864912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/864912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/864912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/864912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/864912/1"}}},"commit":{"parents":[{"commit":"677c89c23631e9083261a1a18ed438d8966e0de2","subject":"Fix set_property_atomic() boolean type casting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/677c89c23631e9083261a1a18ed438d8966e0de2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-09 15:33:37.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-17 15:11:18.000000000","tz":-480},"subject":"Fix unintentional exception inspecting VMDK","message":"Fix unintentional exception inspecting VMDK\n\nIt looks like a raise statement was left in the virtual_size property\nhandler for VMDK, which should have been converted to a log at some\npoint. All the other inspectors return zero for virtual_size if the\nformat does not match or they are unable to parse the data. This\nconverts that raise to a log, and adds a test to make sure we make it\nfar enough in the processing of the complex VMDK format to ensure that\nbehavior.\n\nCloses-Bug: #1983279\n\nChange-Id: I0352ab6b2c00055de094ac5902b8d50941d06dcf\n(cherry picked from commit 199722a65a88f9c28b554b180ec36415ed3a51ac)\n(cherry picked from commit 07f98747a2d4bf65f44348603ecba639d602db4d)\n(cherry picked from commit 323e9f5f93c99121c6ee34e711fb4fd1d191e430)\n(cherry picked from commit 2281c0868e00d0289e4a5c78fbff99f67b5d3d4d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8845a101a30ce010215f0f25f1419d992db7956d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8845a101a30ce010215f0f25f1419d992db7956d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
