)]}'
{"id":"openstack%2Fglance~867880","triplet_id":"openstack%2Fglance~stable%2Fzed~I22e177b4acdc29fe082955cc9950531397037580","project":"openstack/glance","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-12-15 20:19:19.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I22e177b4acdc29fe082955cc9950531397037580","subject":"DNM: testing tox4 and pinning on stable branch","status":"ABANDONED","created":"2022-12-15 19:17:34.000000000","updated":"2022-12-15 20:35:23.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"cherry_pick_of_change":867893,"cherry_pick_of_patch_set":1,"meta_rev_id":"d7f4d6f40322681b8471c16a5e70dacb9ea25b68","_number":867880,"virtual_id_number":867880,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-15 20:10:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-15 19:18:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a2d38b52926ad1bc5bcb922ac02b6542bb79e98","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 19:17:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"af2ba707605aaee336838a6137829d770ed428eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:18:05.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch stable/zed.  The error was:\n\n  while scanning a simple key\n  in \"openstack/glance/.zuul.yaml@stable/zed\", line 305, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/glance/.zuul.yaml@stable/zed\", line 306, column 40","accounts_in_message":[],"_revision_number":1},{"id":"355b6aded92454ae37e5bd44c98d3cb56ec68d5e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 19:18:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0415c65662a8fee721441a926be8f691e0412a46","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 19:30:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a995527ef9a3862a3aad49ec99829048d82214a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 19:30:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"810a6cb3d633a239c816516097e663ff9ab53aab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:43:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98f34caf24e94237bbedd9b708356605\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d260026e9f4fbb8b2a35f8956be00a : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc22c3f3cade4c0cb985590f12ac58f5 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39efb97b2fc9471d8e9b621b5c4ab6d0 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af894089f74242bcab37c91c118c4495 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c3c12eab7154973bd152968e4687edd : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/914f57bf4b074a4dae778246c1eb69f1 : SUCCESS in 8m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d8776088ec8f43e0a908bab6d583bcab : SUCCESS in 10m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0ce6191105bc46e7accbc4a02e2aba82 : FAILURE in 9m 53s","accounts_in_message":[],"_revision_number":4},{"id":"bb750c7ad00f538ff3c65dc57448d0ffa51a6874","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 19:51:14.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"842b4e67a97a193b5dadd7cc85b724be7549009f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 19:57:14.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"28b61bf250691b1a346aa7d13b166d6810a52946","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 20:10:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b4aa5678ff9456aaea5465100afa5c5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9a88b4b1497c43d38872e0d183657b35 : SUCCESS in 3m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/907c7f2f454245ecb83cedbe41ce68e6 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70882fa2cddb47dbb4a0c5717400a06e : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbd9463f4bb349c498a1fcdd518f7ab5 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d31714663bc4f6d8064acac7d472780 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2e331ae68b141a3b392a9c7a338b7fe : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c609b8873b1d44a09d5ba0ce3cf6059a : FAILURE in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3c7b8404d764a14a13d9ddab2bfb111 : SUCCESS in 5m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d4da94feccb14346b90f9c129599c470 : SUCCESS in 10m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c8bc0d4ac6ce4ec8a16121b9d8d69efe : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":5},{"id":"299af805d2718e6a0facb907e9764476bf7ac816","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 20:19:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d7f4d6f40322681b8471c16a5e70dacb9ea25b68","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 20:35:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6c389f89a7b5dbf3b2ae66135e427abd38d0c17b","revisions":{"a32e24dacbaeaeb042d3eae5b1aa40e91c27d123":{"kind":"REWORK","_number":1,"created":"2022-12-15 19:17:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/80/867880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/867880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/867880/1"}}},"commit":{"parents":[{"commit":"07f98747a2d4bf65f44348603ecba639d602db4d","subject":"Fix unintentional exception inspecting VMDK","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07f98747a2d4bf65f44348603ecba639d602db4d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:16:19.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:17:34.000000000","tz":0},"subject":"DNM: testing tox4 and pinning on stable branch","message":"DNM: testing tox4 and pinning on stable branch\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849\nChange-Id: I22e177b4acdc29fe082955cc9950531397037580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a32e24dacbaeaeb042d3eae5b1aa40e91c27d123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a32e24dacbaeaeb042d3eae5b1aa40e91c27d123"}]},"branch":"refs/heads/stable/zed"},"7ff9737c471578aef652511812b202334cbc098d":{"kind":"REWORK","_number":2,"created":"2022-12-15 19:18:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/80/867880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/867880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/867880/2"}}},"commit":{"parents":[{"commit":"07f98747a2d4bf65f44348603ecba639d602db4d","subject":"Fix unintentional exception inspecting VMDK","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07f98747a2d4bf65f44348603ecba639d602db4d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:16:19.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:18:56.000000000","tz":0},"subject":"DNM: testing tox4 and pinning on stable branch","message":"DNM: testing tox4 and pinning on stable branch\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849\nChange-Id: I22e177b4acdc29fe082955cc9950531397037580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ff9737c471578aef652511812b202334cbc098d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ff9737c471578aef652511812b202334cbc098d"}]},"branch":"refs/heads/stable/zed"},"241626524aca85be77e847d87b4e25efaf8c26ae":{"kind":"REWORK","_number":3,"created":"2022-12-15 19:30:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/80/867880/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/867880/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/867880/3"}}},"commit":{"parents":[{"commit":"07f98747a2d4bf65f44348603ecba639d602db4d","subject":"Fix unintentional exception inspecting VMDK","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07f98747a2d4bf65f44348603ecba639d602db4d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:29:38.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:29:38.000000000","tz":-360},"subject":"DNM: testing tox4 and pinning on stable branch","message":"DNM: testing tox4 and pinning on stable branch\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849\nChange-Id: I22e177b4acdc29fe082955cc9950531397037580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/241626524aca85be77e847d87b4e25efaf8c26ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/241626524aca85be77e847d87b4e25efaf8c26ae"}]},"branch":"refs/heads/stable/zed"},"c29b660d3a03818d06adee6c62fd75c4ce801f5c":{"kind":"REWORK","_number":4,"created":"2022-12-15 19:30:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/80/867880/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/867880/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/867880/4"}}},"commit":{"parents":[{"commit":"07f98747a2d4bf65f44348603ecba639d602db4d","subject":"Fix unintentional exception inspecting VMDK","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07f98747a2d4bf65f44348603ecba639d602db4d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:29:38.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:30:50.000000000","tz":-360},"subject":"DNM: testing tox4 and pinning on stable branch","message":"DNM: testing tox4 and pinning on stable branch\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849\nChange-Id: I22e177b4acdc29fe082955cc9950531397037580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c29b660d3a03818d06adee6c62fd75c4ce801f5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c29b660d3a03818d06adee6c62fd75c4ce801f5c"}]},"branch":"refs/heads/stable/zed"},"6c389f89a7b5dbf3b2ae66135e427abd38d0c17b":{"kind":"REWORK","_number":5,"created":"2022-12-15 19:57:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/80/867880/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/867880/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/867880/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/867880/5"}}},"commit":{"parents":[{"commit":"07f98747a2d4bf65f44348603ecba639d602db4d","subject":"Fix unintentional exception inspecting VMDK","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07f98747a2d4bf65f44348603ecba639d602db4d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:29:38.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:57:14.000000000","tz":0},"subject":"DNM: testing tox4 and pinning on stable branch","message":"DNM: testing tox4 and pinning on stable branch\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849\nChange-Id: I22e177b4acdc29fe082955cc9950531397037580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c389f89a7b5dbf3b2ae66135e427abd38d0c17b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c389f89a7b5dbf3b2ae66135e427abd38d0c17b"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
