)]}'
{"id":"openstack%2Fglance~867893","triplet_id":"openstack%2Fglance~master~I22e177b4acdc29fe082955cc9950531397037580","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-12-15 19:49:15.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I22e177b4acdc29fe082955cc9950531397037580","subject":"DNM: testing tox4 and pinning on stable branch","status":"ABANDONED","created":"2022-12-15 19:16:46.000000000","updated":"2022-12-21 18:28:31.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"babee99efc0e6d338dec43beaae76ce2525b384d","_number":867893,"virtual_id_number":867893,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-15 19:42:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-15 19:17:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cedc52250bb38df8b71ba597858e0df694569662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 19:16:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cc1244ece591e5f544a83bda622168d72b9ee74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:17:20.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/glance/.zuul.yaml@master\", line 360, column 31","accounts_in_message":[],"_revision_number":1},{"id":"1217a22af08b7d071f7ea1c872c2f975a3c6e7be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:17:21.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/glance/.zuul.yaml@master\", line 360, column 31","accounts_in_message":[],"_revision_number":1},{"id":"d90c3d08c7193278228323fcf86d1a9e8b2fca56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 19:20:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"34343cc175cfa2fbfd5648d9b18a66c3cf8530ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:21:23.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/glance/.zuul.yaml@master\", line 325, column 13\ndid not find expected key\n  in \"openstack/glance/.zuul.yaml@master\", line 335, column 14","accounts_in_message":[],"_revision_number":2},{"id":"b4f430cd4bfc527ecc5ae64f7df5fd2f93a3a001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:21:28.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/glance/.zuul.yaml@master\", line 325, column 13\ndid not find expected key\n  in \"openstack/glance/.zuul.yaml@master\", line 335, column 14","accounts_in_message":[],"_revision_number":2},{"id":"5718f58ba96bee1ba53b739ed7c07bac32edd89b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 19:25:37.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7f30dce8c224ccc20a5896c8a787a688794fb723","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:26:14.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/glance/.zuul.yaml@master\", line 324, column 13\ndid not find expected key\n  in \"openstack/glance/.zuul.yaml@master\", line 334, column 14","accounts_in_message":[],"_revision_number":3},{"id":"c213935584d6593fd2f0b50bed6f43832a8bc05a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:26:21.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/glance/.zuul.yaml@master\", line 324, column 13\ndid not find expected key\n  in \"openstack/glance/.zuul.yaml@master\", line 334, column 14","accounts_in_message":[],"_revision_number":3},{"id":"2c5666eabf3cfbd40b9fddf9b931cbacc161d361","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 19:26:33.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"52bdda5146b550a93281a780afb692f5aca59422","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:36:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bf0a28e2338410fbac949999281db17\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/09f036deb4a34f96b0ff1c08de3403a5 : FAILURE in 7m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59b4cf838c5c40e2ab117431c527c9e1 : FAILURE in 7m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5c831972da3490b55102d85d9f8346c6e6a500c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:42:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d980393b49394f8d86089b0a6ab3c91c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4b0dc2467b86455e9b348f0f07390521 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ed9167a0365426894a1a0cfec87283e : FAILURE in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07f84f727f024bef9cae4b6280bb09f5 : FAILURE in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18298a0357e147569eed799eb86bd5da : FAILURE in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64bb4d1194604fb2ae7056e69291c866 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba7d709b64a745798d62c2011fc4b5f0 : SUCCESS in 4m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4f368896dbd24b71b0cfa635edba10ca : SUCCESS in 11m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/406a6ed723024bcf9c2b23868f24b65e : FAILURE in 2m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/038cb4bab6744124928225cf00cfde57 : FAILURE in 3m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b68114351c6c42b8b89d8c7a7fbcdf76a6ec234c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-15 19:49:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"babee99efc0e6d338dec43beaae76ce2525b384d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-21 18:28:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4cd48d4440ff2a58712025fe9923ffdb201977a1","revisions":{"54954d7a8a1460d6b734b61e51103ef7c5ea2521":{"kind":"REWORK","_number":1,"created":"2022-12-15 19:16:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/867893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/867893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/867893/1"}}},"commit":{"parents":[{"commit":"bc9856ea9ce2ced219e9b294d8db3b5645d69796","subject":"Merge \"Quota configuration: improve example oslo_limit section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc9856ea9ce2ced219e9b294d8db3b5645d69796"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:16:19.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:16:19.000000000","tz":-360},"subject":"DNM: testing tox4 and pinning on stable branch","message":"DNM: testing tox4 and pinning on stable branch\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849\nChange-Id: I22e177b4acdc29fe082955cc9950531397037580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54954d7a8a1460d6b734b61e51103ef7c5ea2521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54954d7a8a1460d6b734b61e51103ef7c5ea2521"}]},"branch":"refs/heads/master"},"7ccb5908def4b7c658598d4305dee1af6cafdce4":{"kind":"REWORK","_number":2,"created":"2022-12-15 19:20:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/867893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/867893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/867893/2"}}},"commit":{"parents":[{"commit":"bc9856ea9ce2ced219e9b294d8db3b5645d69796","subject":"Merge \"Quota configuration: improve example oslo_limit section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc9856ea9ce2ced219e9b294d8db3b5645d69796"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:16:19.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:20:27.000000000","tz":0},"subject":"DNM: testing tox4 and pinning on stable branch","message":"DNM: testing tox4 and pinning on stable branch\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849\nChange-Id: I22e177b4acdc29fe082955cc9950531397037580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ccb5908def4b7c658598d4305dee1af6cafdce4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ccb5908def4b7c658598d4305dee1af6cafdce4"}]},"branch":"refs/heads/master"},"72ebee8ddc55fa7a322988107bef92dc0c90a4e9":{"kind":"REWORK","_number":3,"created":"2022-12-15 19:25:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/867893/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/867893/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/867893/3"}}},"commit":{"parents":[{"commit":"bc9856ea9ce2ced219e9b294d8db3b5645d69796","subject":"Merge \"Quota configuration: improve example oslo_limit section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc9856ea9ce2ced219e9b294d8db3b5645d69796"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:16:19.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:25:37.000000000","tz":0},"subject":"DNM: testing tox4 and pinning on stable branch","message":"DNM: testing tox4 and pinning on stable branch\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849\nChange-Id: I22e177b4acdc29fe082955cc9950531397037580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/72ebee8ddc55fa7a322988107bef92dc0c90a4e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/72ebee8ddc55fa7a322988107bef92dc0c90a4e9"}]},"branch":"refs/heads/master"},"4cd48d4440ff2a58712025fe9923ffdb201977a1":{"kind":"REWORK","_number":4,"created":"2022-12-15 19:26:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/867893/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/867893/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/867893/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/867893/4"}}},"commit":{"parents":[{"commit":"bc9856ea9ce2ced219e9b294d8db3b5645d69796","subject":"Merge \"Quota configuration: improve example oslo_limit section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc9856ea9ce2ced219e9b294d8db3b5645d69796"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:16:19.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-12-15 19:26:33.000000000","tz":0},"subject":"DNM: testing tox4 and pinning on stable branch","message":"DNM: testing tox4 and pinning on stable branch\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849\nChange-Id: I22e177b4acdc29fe082955cc9950531397037580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4cd48d4440ff2a58712025fe9923ffdb201977a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4cd48d4440ff2a58712025fe9923ffdb201977a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
