)]}'
{"id":"openstack%2Fglance~868530","triplet_id":"openstack%2Fglance~master~I0a377c9329cef8b251b800018fc0d7f784008329","project":"openstack/glance","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-12-28 06:19:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0a377c9329cef8b251b800018fc0d7f784008329","subject":"Fix tox4 error","status":"MERGED","created":"2022-12-25 03:27:05.000000000","updated":"2022-12-28 06:20:44.000000000","submitted":"2022-12-28 06:19:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868530-tox4","meta_rev_id":"41a98fd9ab2e7898c1de741f854966ee95ee98f4","_number":868530,"virtual_id_number":868530,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-28 06:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-12-28 05:27:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-12-28 05:27:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-25 03:44:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-25 04:14:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-28 05:27:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"7efe2886e32e366d816ae8ba4eca8d506d925a7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-25 03:27:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37c896dc783dd526bf6d9aaf5ea5c61290619642","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-25 03:44:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea0c299581384c9f897ef7581254efef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a116917724c47f1a2a4e1608f06d909 : FAILURE in 15m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81332777c08242ca8738708883f3baff : FAILURE in 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b703fd26c6b3eb64658f6097ac56fa2d6f952d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-25 04:14:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/649318ccb5c049a4b439a73200fa176c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e029a3c734244f67b1bb7ea75929bacf : SUCCESS in 4m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/760e828288e34f679d21350701fc620d : SUCCESS in 46m 09s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/deee07ac531d464daf178a9dc5bad058 : FAILURE in 24m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d5ddfb816ce474fbe993d633d0e8b03 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/117912e6ba2f412d98760d7ee709e18a : FAILURE in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c6e2a48a2894b1eb2553fc6a0e341e5 : FAILURE in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc850e7bd004422d8fa481a542ab6ca5 : FAILURE in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9af9d6733e8047bb9855967d449fd377 : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/aab02ae3537c4fcaa54fa757a3105258 : SUCCESS in 11m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/69b0ef76a7ef4aae94e5de5e4b5b690a : FAILURE in 13m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a95c37a1f297458b869041a79ad7251f : FAILURE in 13m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c19d9234f6594672bcaa1d620817d797 : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a31c0f3ab63f335bbcfe9cde0d9a6152afa461f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-26 00:23:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e37f556c83291b17f4c37838dd73f2cdbed1e098","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 00:37:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fcc0250f747472a86a7e81fd6c56927\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d43ca164ec4349b4bd0f7556831da236 : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59e8235b35c9441298fbcc065a645e4e : SUCCESS in 11m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1237c4daeb84fd812a4efb3d0a2502e5dbd0dfd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 01:21:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19690443c2774eceb03ac53e710afeac\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/370158e7a74e4bfb964abf41ca40cd4f : SUCCESS in 4m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/22e1b412499844c6bb32aa56b8cbb999 : SUCCESS in 56m 25s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/154c3a4631eb474dbe2140fdbaae35db : FAILURE in 29m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e1a8e7b6654404906d6e7c6c3c1e5f : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f547e7468014d4282404fccd4aca43c : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b23be2107e6c4c96a69f28d92c7021bc : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e996fe4d23c546cdbe341d2dcc11ba08 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e13a1b47d59f47ef9c8c428fd93f0367 : SUCCESS in 5m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/df1e739ad4d84d6691341815f7e54e88 : SUCCESS in 10m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1c4dfc2867b14f449960d74f05f26d8f : SUCCESS in 7m 49s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/eb472c0e736b43a78371758020685fb3 : SUCCESS in 8m 16s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2f01e4c24e12430390216f23a7b93e74 : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":2},{"id":"fbf24e0bac1075a1433be7252342e407ea505470","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-26 19:49:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84c51fac1dcc90089907cd7f5c7ce854588b6f98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 20:01:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e69359be0cf4db6990ab98c3bad3b5c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/34a4d8294c1d411fb88242fb21dcf180 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36c29ce1ecda4484b401cc62b26a492c : SUCCESS in 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76b2ee5dc5a605a5f68d7e9d25030dd5557ea2f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 20:31:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/592d946c672d4a2db74a9c419dea8a59\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4b27592ca5104964bf33aa65c391ee85 : SUCCESS in 4m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/199b63a89a0345768e7e4c525fcb2f82 : SUCCESS in 41m 38s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d105a85729a4138a663573fe58ed9a6 : SUCCESS in 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f4ef276dbd4ce4930385b268acf5fc : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7d6771e6910467b9b07d5a4d07701c9 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27458c9ac4734f5eb9fae2c2f2737354 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c97afb600e844fcd85e4c448beb1339e : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/483e64a54c7c487393c570c4601635f7 : SUCCESS in 6m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/13d95bd8acea4d69aff14cc669f5e0dc : SUCCESS in 11m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/41af2d1d56214f1ca7a6ecf8a39f04b1 : SUCCESS in 7m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b154d6f26c994e468eaab3fdb08194f6 : SUCCESS in 6m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ca4222c88a6140748c44f34085a965f4 : SUCCESS in 27m 06s","accounts_in_message":[],"_revision_number":2},{"id":"f857bfe05f051f116d51656aec7cff1e3f707969","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-12-28 05:27:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fdff51a1edcd6cd868734149685ee4909fc79897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 05:27:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b332c4bbb90060c85f549e101876ffec3effc828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 06:19:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae3aa2f6a7884a95a077818e46b65704\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6323486a0e7a426ca5875d4edb78928f : SUCCESS in 3m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a94a1d82c3a427c9334c5c79b2a4fc1 : SUCCESS in 49m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccef68c73c0b42498c9f047654c6222e : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaad9029bf864481afa38fbb50e8bf02 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bea7eb92b56d4c6b92b7b88b9cd9c295 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b77cb14d0df4a4fbb39a06081d54083 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a10ca0f03244f22a259557bde2db5ce : SUCCESS in 6m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6a62d4b9084b449f8bbaeca49658c454 : SUCCESS in 7m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09e56261e03e450580cdc15e84fad8a4 : SUCCESS in 8m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b67bf0acedb402580619f9abaf0a3b9 : SUCCESS in 43m 39s","accounts_in_message":[],"_revision_number":2},{"id":"eb6aa062de0c8b23d9485720c84ba33525fa571f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 06:19:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"41a98fd9ab2e7898c1de741f854966ee95ee98f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 06:20:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eb293c03d1f453c981cce4932f7149a\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a46958ee2f484e4697e7588211a119e4 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/791f8365171245839d35d91f4553534d : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/702cfd6e71044ac787c14b9bc7262895 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2cd7251f46e9d8fe7796ba0e178ee692de98fe6","revisions":{"9e2bd3a6c1680fd609ccbf7be43bbb0212f719d2":{"kind":"REWORK","_number":1,"created":"2022-12-25 03:27:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/30/868530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/868530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/868530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/868530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/868530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/868530/1"}}},"commit":{"parents":[{"commit":"bc9856ea9ce2ced219e9b294d8db3b5645d69796","subject":"Merge \"Quota configuration: improve example oslo_limit section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc9856ea9ce2ced219e9b294d8db3b5645d69796"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 03:26:39.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 03:27:03.000000000","tz":-480},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: I0a377c9329cef8b251b800018fc0d7f784008329\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e2bd3a6c1680fd609ccbf7be43bbb0212f719d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e2bd3a6c1680fd609ccbf7be43bbb0212f719d2"}]},"branch":"refs/heads/master"},"e2cd7251f46e9d8fe7796ba0e178ee692de98fe6":{"kind":"REWORK","_number":2,"created":"2022-12-26 00:23:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/30/868530/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/868530/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/868530/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/868530/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/868530/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/868530/2"}}},"commit":{"parents":[{"commit":"bc9856ea9ce2ced219e9b294d8db3b5645d69796","subject":"Merge \"Quota configuration: improve example oslo_limit section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc9856ea9ce2ced219e9b294d8db3b5645d69796"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 03:26:39.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-26 00:23:59.000000000","tz":0},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: I0a377c9329cef8b251b800018fc0d7f784008329\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2cd7251f46e9d8fe7796ba0e178ee692de98fe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2cd7251f46e9d8fe7796ba0e178ee692de98fe6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
