)]}'
{"id":"openstack%2Fglance~868714","triplet_id":"openstack%2Fglance~stable%2Fyoga~I0a377c9329cef8b251b800018fc0d7f784008329","project":"openstack/glance","branch":"stable/yoga","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-12-28 09:00:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0a377c9329cef8b251b800018fc0d7f784008329","subject":"Fix tox4 error","status":"MERGED","created":"2022-12-27 19:38:38.000000000","updated":"2022-12-28 09:01:10.000000000","submitted":"2022-12-28 09:00:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868714-tox4","meta_rev_id":"b34ce2e231b1613bd66fcfa0ba6e9dca0c734246","_number":868714,"virtual_id_number":868714,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-28 09:00:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-12-28 08:04:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-12-28 08:04:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-27 20:35:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-28 08:04:41.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"456c29e2cdcd1320c6da27ecc26e4b37d05b9f7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-27 19:38:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"355ba230de64cac645c870de77593ac1c4ba01e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 20:35:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ae5ea1c05a340468918e04fe4ce680b\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/247ab27170d3456fb9c141ad7fefe4b3 : SUCCESS in 54m 56s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd42e53c20a7433092e0985be7599653 : SUCCESS in 42m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e225526a43ad4530a472f72f5a6bc88c : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ec77ffd04234cf1838bb6d76aca231a : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6af4525499247678f5b2f237e1d80fd : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b7b900455a4712b3c2c9e86d8e656c : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac59a400f3df418bac1b70e1e9762e7f : SUCCESS in 5m 48s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/7b58da4800e742a699d59a4193a127a3 : SUCCESS in 13m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f484993830164e38a846be4c4fd5dd2b : SUCCESS in 7m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1a4d30b0577e40ef9ab483337c3ea972 : SUCCESS in 10m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c1d987a8098b4f91995e0cb00975a7f9 : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8bcbaa5eab77f58c4aab9e90e607fc7234d8b706","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-12-28 08:04:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f76fd5845b4b2b11a1bb1db45d1a82f28d00b668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 08:05:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01c5d60bad9a20602967dd5224a34ae1752e9d57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 09:00:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd54bcee1bf0423dacb9ba19472027db\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b12f05773bb4e93a3124eb263fec50f : SUCCESS in 49m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2921a94405d24d3d809142440d1a2001 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98e2da6db19e4afea4101a9610507c5f : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a1af1f14652484d90feb88d95738dfc : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f20d19f514f47f78a9bfe834bdd17fc : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/774178a9fc0e4bd6847a49edfadebc6b : SUCCESS in 4m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/974902ba838d4262b87bbd1f5c2ea5c7 : SUCCESS in 8m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f043446e5eea4cdead954d560173decc : SUCCESS in 7m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/47f3136473164a0bb3623e737f0abe68 : SUCCESS in 23m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6968845fc4905adef6dbcf08c7be372170f4ec56","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 09:00:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b34ce2e231b1613bd66fcfa0ba6e9dca0c734246","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 09:01:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/566948d5a38a47f8a49645e3177be41d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/289c418a09e242af9bdc7d364134b487 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f48a4527da34b2d8170cdc9d8474399 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f6c689b7241bc20bc9f7ed398058bc312f81df3","revisions":{"3f6c689b7241bc20bc9f7ed398058bc312f81df3":{"kind":"REWORK","_number":1,"created":"2022-12-27 19:38:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/14/868714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/868714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/868714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/868714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/868714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/868714/1"}}},"commit":{"parents":[{"commit":"5fbf47378e70537c73394408e69e94449aba1bec","subject":"[stable-only] Remove glance-code-constants-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5fbf47378e70537c73394408e69e94449aba1bec"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 03:26:39.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-27 19:38:19.000000000","tz":-360},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: I0a377c9329cef8b251b800018fc0d7f784008329\n(cherry picked from commit e2cd7251f46e9d8fe7796ba0e178ee692de98fe6)\n(cherry picked from commit 022e15025270ea2660dec2d29446658bd8ea9dab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f6c689b7241bc20bc9f7ed398058bc312f81df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f6c689b7241bc20bc9f7ed398058bc312f81df3"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
