)]}'
{"id":"openstack%2Fglance~871831","triplet_id":"openstack%2Fglance~master~I3ec5a33df20e1cfb6673f4ff1c7c91aacd065532","project":"openstack/glance","branch":"master","topic":"fix_format_inspector_vmdk","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-02-06 10:59:35.000000000","reason":"Change was submitted"},"9642":{"account":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"last_update":"2023-02-06 10:59:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ec5a33df20e1cfb6673f4ff1c7c91aacd065532","subject":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX","status":"MERGED","created":"2023-01-26 16:04:13.000000000","updated":"2023-02-06 18:29:07.000000000","submitted":"2023-02-06 10:59:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871831-fix_format_inspector_vmdk","meta_rev_id":"2ef630924e1b3dfbdefd2fef470ee1be1def4faa","_number":871831,"virtual_id_number":871831,"owner":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-06 10:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-02-06 07:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-02-03 16:04:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-02-06 08:57:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-26 16:14:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-26 17:55:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-26 19:00:50.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2023-01-26 19:01:09.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-02-02 05:45:14.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"f1b44333454901053df2673c4979db39f36dd23f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"date":"2023-01-26 16:04:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b7276905ee7e909b033f55d42ce75bda832f3ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 16:14:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/376dd7f3911749e385ea415e9323252e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b602e730d3548f8aae05b28f37b7ac8 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9e3e530307a848ea929e339622930a7b : SUCCESS in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77a8e86134be1752b992af28c1e2c5fe085ea7c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 17:55:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e66c4b369e2455da8bb357b245fcb9a\n\n- grenade https://zuul.opendev.org/t/openstack/build/3f61a470fd4a40faa27f14da0c770fda : SUCCESS in 58m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c7bdf4bf68a34bfca438e60df5225ad5 : SUCCESS in 57m 37s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5813f1d74a3f466c92fd999ceebb88a2 : SUCCESS in 1h 40m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81d50bfbb9574ea2b64eb3db7113c412 : SUCCESS in 53m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b717446d8d61469f86f7008b9244d022 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18ceffac622641059efa78f95fa9cfd0 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/565b5533b8d54b668ae3dce3ebfce319 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cca9fa047ad24a8c91ab6955155d25be : SUCCESS in 5m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3019ccc969de4b469c15d803f4ad58ed : SUCCESS in 10m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b79dacc2ff14bef8bc2a200068ed7f0 : SUCCESS in 9m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/64d78e0e25e44695be16de102f699aa4 : SUCCESS in 7m 51s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4b2b45d7e9e6484d81e526ac0a81f335 : SUCCESS in 1h 11m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c606f087b35d493fae0c1cc53a93fa7f : SUCCESS in 1h 37m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/21a81c7616dc47f88b474ce577f43d03 : SUCCESS in 1h 34m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4276e594d906470d98f3807ebf249311 : FAILURE in 1h 17m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ae6cba1ea1b4db48d7ec9950eed1d28 : SUCCESS in 49m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/654897ba169f4c7792c8d7568853f53f : SUCCESS in 1h 13m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e16ffbd97a90425aa50e017d10078050 : SUCCESS in 34m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a26f82a8405e7634d87014eb9baacf7ac9cf8d62","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-26 19:00:50.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"af46d6d1d3b5d22eda5de2ca0d8dcb43ddfc5ee7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-26 19:01:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bd3f053d757630336a2dc6577daf952e437e8e3","author":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"date":"2023-01-27 09:16:32.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"22c26f29cef2c6ef0ad2e847afd169617ce0e5c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"date":"2023-01-27 11:24:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"594c3c02c664d306bdada369da87ea17637190f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 11:42:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6452907b0f74c85ad6bb52905917d3b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e61f040fbb294f16a62a5611e3ea3ffa : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4fd6c7b8ba474e1c8fee786a8cb6341b : SUCCESS in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01c6dbd9388f4044d90d0c69971644eb704afe15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 13:19:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dac7e37b58054db48b3e8403a308ba29\n\n- grenade https://zuul.opendev.org/t/openstack/build/f874000b84c840b89ca65f96e35f7386 : SUCCESS in 58m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1a88d2188a764fdbba4837da89bdf41f : SUCCESS in 1h 00m 18s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af91a00120aa4b95b7cef8a9cb119d0e : SUCCESS in 1h 37m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aaeee7f5284f440493fc3f52a3152223 : SUCCESS in 56m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c71823ec93b478481e382d0398e849a : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9582746e65354cc093fa16a709113a9e : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45a299166c1c4e96b279fa8dad29333f : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7798b32867491f9ef08766511976fa : SUCCESS in 5m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6812e2c2de854584a169eb79bda39c5d : SUCCESS in 11m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fbb1c9d9c9444058b6ee55ae1ac478ea : SUCCESS in 8m 59s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a1d14e92995e47728ede2f79d0a4a57c : SUCCESS in 8m 02s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6dffdc054c7a42b3b2aec77fe27cd776 : SUCCESS in 1h 20m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fbac165f8bdf486fa6510783b2a7d1d8 : SUCCESS in 1h 30m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/73fdb641989d4cd682c85bc14367a708 : SUCCESS in 1h 14m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/369d8bf5775b4a138a20e5bc33043358 : SUCCESS in 1h 45m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d393bd66ba324a728ad353c55f61ff62 : SUCCESS in 53m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/15f523255be74897a0d31dd9f7698be7 : SUCCESS in 1h 10m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/49fe0858de2b49edbc4442896241fc21 : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":2},{"id":"4e961b7252338f0c9b4bda7a27f0dc11d9f88a48","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-02 05:45:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9928d0b8639f3445cd1dec3b34f85e1fdeaee88","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-02 14:53:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c82453ce40ce1006babc823d32f209fee2c1c66c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-02 14:57:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"de6caeacd0f20f9622ce7c9f0caf9be3f47139e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-03 16:02:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"60c74bd6852aba698f666cad36a17f9ef8207d94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-03 16:04:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e3ef2457cedb2fbc6a45efd771cfbf65e6f952c4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-03 16:04:41.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aa6e9b5bb24227cd1600d4b18325d3db4125f9e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 16:18:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d50a524df074203bc94a2f730bc79d0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e2f7c3279954b52926178a0ca062d98 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/83504806801d4233a4d2ef02b736f5de : SUCCESS in 11m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d5ff2db23a56441cddae25c29c37d5cb8d7919ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 18:05:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47cb11e0e0cf492ab1eb676ed7bf6a48\n\n- grenade https://zuul.opendev.org/t/openstack/build/eb00267e374a401ca153bb49adb44220 : FAILURE in 39m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cb2c6ec45e534608a559fe495a853722 : SUCCESS in 52m 54s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/949efaed8b3e4779bc642423da91436a : SUCCESS in 1h 20m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/442843903c0642a582240bf5aecaed67 : SUCCESS in 51m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/306a53aef8c8494cbf0d93e42722494a : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae733e66dc8249af9637e02eae27315b : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/919072d478f64f249731be8b03a487df : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeddb1a8985a496f9f0a98b9ce6409ad : SUCCESS in 6m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c75dca1853484d548ff6923b1eb20648 : SUCCESS in 13m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0b8b6819991b4e269052da7cfc885a0c : SUCCESS in 8m 30s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a1ace91aa7ee4c19b6a52d37a9250bc9 : SUCCESS in 9m 39s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8df1406fefe34871a90159de06ec0627 : SUCCESS in 1h 06m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f8cc76c336334f0d96f2e9243755d962 : SUCCESS in 1h 49m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d3b920e8db0643949556103acf045b5e : SUCCESS in 1h 33m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cb9ad88f17704d9c819dc1c3ab32bb33 : SUCCESS in 2h 00m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eab0675f23da45d5948eaa2175afe8c0 : SUCCESS in 1h 10m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eefca474d7634b3b9d7dbb37b1213dda : SUCCESS in 1h 21m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/898d946c4a4a456fa8d8782aa61b24d4 : SUCCESS in 26m 40s","accounts_in_message":[],"_revision_number":4},{"id":"6fd6de92c75ee3886761ae7fb96dd0fcf9b268c4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-06 07:05:59.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ef9b8c0dc20a7bbbbef67a887623d5fd500627f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 07:11:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be66022cee8348dcaa1dfa560618a8ff\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72aefd7d08ea4dd2936655c4c7e69990 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af7017bfdc324b539200a10b99735ab9 : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c338438ba21af997d511e3d3ded56a2e788c036e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 08:55:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/379e3f2a14434c6da35350b2427c4b11\n\n- grenade https://zuul.opendev.org/t/openstack/build/cc266a6e439949099269f297967d2975 : SUCCESS in 54m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/03b75551e0504f08944b3dfe58e8318e : SUCCESS in 54m 38s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8f86b3187804442a4a642c7579d78db : SUCCESS in 1h 26m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95ac0695181d4a5688906dc3edbdda6e : SUCCESS in 53m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8989b7a70c51423d9933579488cd5d7b : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7283bf1684142e9b456416c25334425 : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2cac27e6a2dc4cefb9dd2d9f7fb928aa : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ba40f3341dd474cafe4bc435730f14f : SUCCESS in 6m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ea7159e9d84a41bb9eb9778965f27b35 : SUCCESS in 10m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/de255173a9eb4e2e8daa1e490401bb8b : SUCCESS in 10m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/265213caf202451f949f9d6480e7e637 : SUCCESS in 9m 48s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0343019e7ec74fc6a8427f11b22fcfb3 : SUCCESS in 1h 20m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/69e9ba4b326e46ffbedc8ebeb07f82b2 : SUCCESS in 1h 47m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/143c47d4403b46d886632083f5573738 : SUCCESS in 1h 13m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/087b2d7fbad145108db44a6729ce2f97 : SUCCESS in 1h 29m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5774dd9977f342708ef8057981d945db : SUCCESS in 50m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/24d9a2f842924fc4b209e237bdb6251a : SUCCESS in 57m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ad14d88a9e1349989b18b727b2f07334 : SUCCESS in 23m 27s","accounts_in_message":[],"_revision_number":4},{"id":"e0db906fe64261d0ffab642309b66e9cca09ff6c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-06 08:57:30.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4557e562d2a114319783d6c7ddc61cffcdf565d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 08:58:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"efe435185aa699785d79280dbbf5c5c0ca49be20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 10:59:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2014a23feb814831b88ae51618df65d2\n\n- grenade https://zuul.opendev.org/t/openstack/build/c228c4c98c4f4e7fbccf20b8277e949b : SUCCESS in 46m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3687338b3db45068b0e043ef32aa437 : SUCCESS in 1h 05m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/502cee3256e24c5cbc3a7689a6b39e9f : SUCCESS in 49m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/014aea9ed5bc4ccfae8e27cc32693cf7 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a8d673028d045a4bd720c180c02e2b4 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d13e5a9eef240959ab41557e2d5fe4d : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b45d82411e4d9591838afd49ae7f10 : SUCCESS in 6m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e386f9e681d74befbbe0dd22b6c496bf : SUCCESS in 8m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7eaefc4b2e5c4b33b65c1ace2cc31a25 : SUCCESS in 11m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7ba7a05527cd408fa6f32ac2f87adeaa : SUCCESS in 1h 28m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fd412a2d1f994b838879d150e92e66f6 : SUCCESS in 1h 53m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5d9d725fed8f45688e429cca03b4712c : SUCCESS in 1h 10m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89222ec7ac674b07b10f97f115020a71 : SUCCESS in 59m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1f015183073049a9a94dc6dc5e686d52 : SUCCESS in 1h 24m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9f8a5931e8664db2be797ccd93e91f48 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":4},{"id":"6a61348d68f2c069585411180056e01b142082e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 10:59:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"06d0f16662cdb2814523e9f8ab223ef907fdee74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 11:00:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/256aecdbe1e2418798bc67b8ec0ddf30\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44fb588aa944440b90ce6d599913868 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4},{"id":"694ad9372f656994aebffed467aaa7fc9310f3a2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-06 18:19:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2ef630924e1b3dfbdefd2fef470ee1be1def4faa","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-06 18:29:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d4d33ee30f303f783c0640cd72acb31b313e1164","revisions":{"249bc83f1745c49ef4d087960bee258598d8662b":{"kind":"REWORK","_number":1,"created":"2023-01-26 16:04:13.000000000","uploader":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"ref":"refs/changes/31/871831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/871831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/871831/1"}}},"commit":{"parents":[{"commit":"907c56265438da43c13bf1a3369d5459b1267e34","subject":"Merge \"Enforce image safety during image_conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/907c56265438da43c13bf1a3369d5459b1267e34"}]}],"author":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-01-25 10:53:09.000000000","tz":60},"committer":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-01-26 16:03:08.000000000","tz":60},"subject":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX","message":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX\n\nVMDK:\nWhen parsing a VMDK file to calculate its size, the format_inspector\ndetermines the location of the Descriptor section by reading two\nuint64 from the headers of the file and uses them to create the\ndescriptor CaptureRegion.\n\nIt would be possible to craft a VMDK file that commands the\nformat_inspector to create a very big CaptureRegion, thus exhausting\nresources on the glance-api process.\n\nThis patch binds the beginning of the descriptor to 0x200 and limits\nthe size of the CaptureRegion to 1MB, similar to how the VMDK\ndescriptor is parsed by qemu.\n\nVHDX:\nIt is a bit more involved, but similar: when looking for the\nVIRTUAL_DISK_SIZE metadata, the format_inspector was creating an\nunbounded CaptureRegion.\n\nIn the same way as it seems to be done in Qemu, we now limit the upper\nbound of this CaptureRegion.\n\nChange-Id: I3ec5a33df20e1cfb6673f4ff1c7c91aacd065532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/249bc83f1745c49ef4d087960bee258598d8662b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/249bc83f1745c49ef4d087960bee258598d8662b"}]},"branch":"refs/heads/master"},"f15e5119ee9197c7ef4304fd0af985ea80fba75d":{"kind":"REWORK","_number":2,"created":"2023-01-27 11:24:54.000000000","uploader":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"ref":"refs/changes/31/871831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/871831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/871831/2"}}},"commit":{"parents":[{"commit":"907c56265438da43c13bf1a3369d5459b1267e34","subject":"Merge \"Enforce image safety during image_conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/907c56265438da43c13bf1a3369d5459b1267e34"}]}],"author":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-01-25 10:53:09.000000000","tz":60},"committer":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-01-27 11:24:46.000000000","tz":60},"subject":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX","message":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX\n\nVMDK:\nWhen parsing a VMDK file to calculate its size, the format_inspector\ndetermines the location of the Descriptor section by reading two\nuint64 from the headers of the file and uses them to create the\ndescriptor CaptureRegion.\n\nIt would be possible to craft a VMDK file that commands the\nformat_inspector to create a very big CaptureRegion, thus exhausting\nresources on the glance-api process.\n\nThis patch binds the beginning of the descriptor to 0x200 and limits\nthe size of the CaptureRegion to 1MB, similar to how the VMDK\ndescriptor is parsed by qemu.\n\nVHDX:\nIt is a bit more involved, but similar: when looking for the\nVIRTUAL_DISK_SIZE metadata, the format_inspector was creating an\nunbounded CaptureRegion.\n\nIn the same way as it seems to be done in Qemu, we now limit the upper\nbound of this CaptureRegion.\n\nChange-Id: I3ec5a33df20e1cfb6673f4ff1c7c91aacd065532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f15e5119ee9197c7ef4304fd0af985ea80fba75d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f15e5119ee9197c7ef4304fd0af985ea80fba75d"}]},"branch":"refs/heads/master"},"58aac992a3a5d7a36497bea2c1ef9c557dc4aace":{"kind":"REWORK","_number":3,"created":"2023-02-03 16:02:08.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/31/871831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/871831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/871831/3"}}},"commit":{"parents":[{"commit":"907c56265438da43c13bf1a3369d5459b1267e34","subject":"Merge \"Enforce image safety during image_conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/907c56265438da43c13bf1a3369d5459b1267e34"}]}],"author":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-01-25 10:53:09.000000000","tz":60},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-02-03 16:01:50.000000000","tz":-480},"subject":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX","message":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX\n\nVMDK:\nWhen parsing a VMDK file to calculate its size, the format_inspector\ndetermines the location of the Descriptor section by reading two\nuint64 from the headers of the file and uses them to create the\ndescriptor CaptureRegion.\n\nIt would be possible to craft a VMDK file that commands the\nformat_inspector to create a very big CaptureRegion, thus exhausting\nresources on the glance-api process.\n\nThis patch binds the beginning of the descriptor to 0x200 and limits\nthe size of the CaptureRegion to 1MB, similar to how the VMDK\ndescriptor is parsed by qemu.\n\nVHDX:\nIt is a bit more involved, but similar: when looking for the\nVIRTUAL_DISK_SIZE metadata, the format_inspector was creating an\nunbounded CaptureRegion.\n\nIn the same way as it seems to be done in Qemu, we now limit the upper\nbound of this CaptureRegion.\n\nChange-Id: I3ec5a33df20e1cfb6673f4ff1c7c91aacd065532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/58aac992a3a5d7a36497bea2c1ef9c557dc4aace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/58aac992a3a5d7a36497bea2c1ef9c557dc4aace"}]},"branch":"refs/heads/master"},"d4d33ee30f303f783c0640cd72acb31b313e1164":{"kind":"REWORK","_number":4,"created":"2023-02-03 16:04:23.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/31/871831/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/871831/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/871831/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/871831/4"}}},"commit":{"parents":[{"commit":"907c56265438da43c13bf1a3369d5459b1267e34","subject":"Merge \"Enforce image safety during image_conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/907c56265438da43c13bf1a3369d5459b1267e34"}]}],"author":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-01-25 10:53:09.000000000","tz":60},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-02-03 16:04:17.000000000","tz":-480},"subject":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX","message":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX\n\nVMDK:\nWhen parsing a VMDK file to calculate its size, the format_inspector\ndetermines the location of the Descriptor section by reading two\nuint64 from the headers of the file and uses them to create the\ndescriptor CaptureRegion.\n\nIt would be possible to craft a VMDK file that commands the\nformat_inspector to create a very big CaptureRegion, thus exhausting\nresources on the glance-api process.\n\nThis patch binds the beginning of the descriptor to 0x200 and limits\nthe size of the CaptureRegion to 1MB, similar to how the VMDK\ndescriptor is parsed by qemu.\n\nVHDX:\nIt is a bit more involved, but similar: when looking for the\nVIRTUAL_DISK_SIZE metadata, the format_inspector was creating an\nunbounded CaptureRegion.\n\nIn the same way as it seems to be done in Qemu, we now limit the upper\nbound of this CaptureRegion.\n\nChange-Id: I3ec5a33df20e1cfb6673f4ff1c7c91aacd065532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4d33ee30f303f783c0640cd72acb31b313e1164"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4d33ee30f303f783c0640cd72acb31b313e1164"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
