)]}'
{"id":"openstack%2Fglance~872522","triplet_id":"openstack%2Fglance~master~I0808dc0b1b34b527e38aa137c1dd25e1fc06409f","project":"openstack/glance","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2023-02-20 15:13:36.000000000","reason":"Change was submitted"},"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2023-02-20 15:13:36.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-02-20 15:13:36.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-02-16 13:52:58.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-02-20 15:13:36.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-02-14 12:59:20.000000000","reason":"removed on reply"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-02-16 18:27:40.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I0808dc0b1b34b527e38aa137c1dd25e1fc06409f","subject":"Enabled new defaults and scope checks by default","status":"MERGED","created":"2023-02-02 10:42:47.000000000","updated":"2023-02-20 15:16:04.000000000","submitted":"2023-02-20 15:13:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":80,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"872522-secure-rbac","meta_rev_id":"1d2a534fe09e1d9e7f11556c86c9e8dc6c71b64f","_number":872522,"virtual_id_number":872522,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-20 15:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"disliked":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-1,"date":"2023-02-16 14:39:33.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2023-02-16 13:52:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2023-02-20 13:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":1,"date":"2023-02-16 18:27:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2023-02-20 13:40:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-02 12:19:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-02 14:00:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-02 19:29:05.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-02-08 09:17:50.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-02-10 15:08:33.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-02-14 02:48:51.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-02-14 02:48:51.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2023-02-14 02:48:51.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"49277fe51197507729692ed074143ef0d3494116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-02 10:42:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b77d2a6775ef7c48b6a46d44fbeb06495b09abd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 12:19:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbf8a95cf7254c3ab9e0e2aac2d8d789\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/146026e2ed2c4970b2ab13d07b29a4eb : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/efd00b195f284552b9d2119f806e7e8f : SUCCESS in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54fd8a2fc463ffd959e5aacf1d0a67378d7ca454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 14:00:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ca2b8706e954964b681a2da27cfeba6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8a4180349f6542b4a1819aeec25ca736 : SUCCESS in 2m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/df2774d0a1c345599c6acb2da01a5840 : FAILURE in 58m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1c830a122feb460c8b9701cdb4a90bed : FAILURE in 54m 25s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5fee5e5d344409a9569762b51579fb4 : SUCCESS in 1h 28m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebe30608fcc1441ab096506f24983328 : SUCCESS in 1h 03m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a2079f2116d447c9822df410a753a07 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c33c04dd7754ea4a445711072790e73 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da03021da0214b8bb633dd7979bca45c : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea404f61fdb484fac7f99f4d7cf1b40 : SUCCESS in 5m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9f7d62050de948b494115902b4072219 : FAILURE in 18m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8df5c695725d448a939553e981d1e7f5 : FAILURE in 12m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/57afc1d930b744cf8b905d36b069d041 : FAILURE in 11m 15s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/de4863f3b9c3451496d1157992b4899a : FAILURE in 1h 23m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/91c47ffba0ef43e9bc876abb89b7efee : SUCCESS in 1h 27m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ea6c086631a247d9869486b11becdb31 : SUCCESS in 1h 29m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cfdb904258b64458b50942c63887df37 : SUCCESS in 1h 51m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/199a8e31788842abaeca67f497d343dd : SUCCESS in 34m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/64c93b0a58224662ad9a53ca46b11924 : SUCCESS in 1h 20m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a7adc297b4ce4a4ba4259b8493f29b96 : SUCCESS in 32m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3b34f3d7e75de36e17a9650728378b1fb880d476","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-02 19:29:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"65dd5c1c4fea7cedaf3dee002ddedc3da13a86d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-08 08:00:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3984d5578bd6e81dc3fb91ff393f3aace6e707d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 08:07:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85c4b59b07da4e1cb0a3832ebc0ced55\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b39d821f5c441d2a45226f9a9447b76 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a6dd585b676745e49f6e3ce8bdd5fec4 : SUCCESS in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c9ee9d8aef7fbfecfd0bc49fa8d290880cf7d25","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-08 08:14:06.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1bbd08c34598906a716fa328945910b79e97a6bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-08 09:16:58.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"41c5b0d65fbe110212d160ab3f95d0ebc1df10fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 09:31:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/223d476a7ac54813878dbaeb788c8631\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb85b5fa172f4920ad9a80c73497fbaf : SUCCESS in 11m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9a4687aaa27a46ca9d38929138d1b52c : SUCCESS in 7m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4ea4ba623dffd5698a387bdaee339ef3e900c5e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 11:34:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbadee09fb7643b7be7b1246f0197c30\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/af93e9a94c004f859c971258cb5b4274 : SUCCESS in 4m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/ee0adcac011f459798a5529dc7368008 : SUCCESS in 45m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1dcdb3c1745f41af97fbff1a9f79deb2 : SUCCESS in 45m 15s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8639d78f4f5b4ec1a50d24fc02d40249 : SUCCESS in 1h 18m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da72b6eac6254c539ae300ebeac59373 : SUCCESS in 56m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9472be88432e4ead81a170d544e8f8c0 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d89eec78a26458daedb2b99c2504b28 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2725c2ca79f042dd903cdf0dd52ab07e : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2403de46e244161b55e78c06b0d4de3 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3292052b2914bf18b9cd5a50defc2f6 : SUCCESS in 4m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/79d01ebd78d841db877deadf7e102500 : SUCCESS in 12m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a593510ea9864476b545bf93e2ce2840 : SUCCESS in 12m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7d673e85e7324ce3b6b1c13261388915 : SUCCESS in 8m 54s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c09c881ad99d4b90813b1b2306717caf : SUCCESS in 1h 12m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f7bb938616284e32a0ef864ac7a7840d : SUCCESS in 1h 32m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2eea0155ea474ff39bb09404348fec50 : SUCCESS in 1h 22m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d8f066b8a5f1475583be7ed83470d853 : SUCCESS in 1h 25m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e339ae98f71244a18bb87b68a40c6c3b : SUCCESS in 1h 33m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9efa560ae2f448b87fcc35d33d61000 : SUCCESS in 45m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/23be20ceda3240f289840a8037bca9d9 : SUCCESS in 1h 28m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8b364f5d4dfb4f618941683d18194822 : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":3},{"id":"4c4903a079f23638c28667ae1df9e0ca06185732","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-09 02:47:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ca1dcc64d4bf855cfadc566c331ad7f668698346","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-09 12:45:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"08511197370ad3f5b3f05cc035ab189492b2f562","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-10 00:44:09.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"02e4ebefac9c4fd7fe7847d51adf4b55ab4990f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-10 14:05:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"b93b28a439852ba4fca3abf37040baf58f2cba16","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-10 14:08:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6b3583aa87df57b0471e9d6af1d3f322ed217316","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 14:24:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d5bdbf16a024a0aa30a2e08ac958db1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01803c98698a41aaa2b04604129806de : SUCCESS in 15m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/11e22b59b2e148bca7a65daa68381748 : SUCCESS in 14m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a5cc0fa1f2c628592702a2d7f1c79a7482e9c88e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-10 14:54:36.000000000","message":"Patch Set 4: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c535f82aca77c2a3732b45b896452e55569354e9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-10 14:57:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c2fc358036e7f73166527f2ac86d5f6faea73b78","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-10 15:08:33.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fa6cf5c70828b03c435cbc2b436ac9413f80b10e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 16:00:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35df2467415846bea229d4d60faa19b4\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/05a5db90dfa841c390c6a27cb0485936 : SUCCESS in 2m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/f8f26b7e3be34a1ca0128725502ff728 : SUCCESS in 48m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/309f0100ebee47b7af6f2060efebe9da : SUCCESS in 1h 16m 34s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a0f3e1d4c9a486997b8ac5da4356bce : SUCCESS in 1h 49m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24f59d918e0f48a9b0311cf3c16e1569 : SUCCESS in 46m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4c3ec8b5a44cff92c446bd57509a70 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eacfe62382854f5e9d589d5cdfce4e47 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c41dbe1ecf824b7798872f2edadd7481 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03d5755e88104bc89c4b5998ef057a94 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cab43863981241ffb9605c0755e3bc63 : SUCCESS in 6m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8f43c1482bf74f40846f95eea1882f5a : RETRY_LIMIT in 5m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aad73d9214314f689b169fec85052476 : SUCCESS in 7m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ff79eba08f2a446ba5964a25d29736df : SUCCESS in 8m 52s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5b78c1d10db749fb88a4c85925e5ffef : SUCCESS in 1h 14m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ffca53c34b3e4ab4a6a479d87fc67579 : SUCCESS in 1h 35m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1f423bb1581d42d8bacd268e2a851a35 : SUCCESS in 1h 32m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3bc44e50afb548f997756856e7b54ffd : SUCCESS in 1h 35m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ef8bd4dfe69847ed885089ee778c1af4 : SUCCESS in 1h 42m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d7bee1880e44b5cadd2644562733280 : SUCCESS in 50m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0e38d27b451b46fc9aadd34b9353b886 : FAILURE in 1h 15m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9b57ce55612e4791884582a7b4047192 : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":4},{"id":"70a5a3a6aea02dd8557f61ebff9884f097d5c9d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-11 11:47:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8b35d2f42a73cd81a1c0012c37d128a4bbf8988c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-11 11:48:12.000000000","message":"Patch Set 5:\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":5},{"id":"526390641ce29fb3fbc1533b911ea8f1c582b12f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-11 11:55:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cf0db61ee970da38b247cf06eff58ce848596ad0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-11 11:56:10.000000000","message":"Patch Set 6:\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":6},{"id":"d30c35c5f5fc28fbee1dc562bcd8f529e764b596","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-11 11:57:52.000000000","message":"Patch Set 6:\n\n(12 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a392ec01f31060a2bc7d4226f0b32b920445c35c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-11 14:24:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/030116a29ba24cfc90b25afcbd39b46f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/06599d7c085c4d4fa70e268380fec4e4 : SUCCESS in 3m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/0bda3fa2388c4f338e66a2bc77c26487 : FAILURE in 1h 46m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6228904c4eeb4057b5cd92a32931f61c : FAILURE in 1h 42m 38s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a2c827538304432ade56cf81bb7cb30 : SUCCESS in 1h 24m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97d8317b9d654598a5977731d8c62b02 : SUCCESS in 1h 48m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28681a43ffaf49189d12c484ae5db0b8 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/414cd62e6329465dae49fc216dbe63ff : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/380650188dc449ca84ceb08fadd46044 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13da05d0687640d8a1453eee6a82b303 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b2e9647ad5f4306a67eea9b81447eb3 : SUCCESS in 7m 36s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bea421a6455844b1bb2f1865e836aa1e : RETRY_LIMIT in 6m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c0c1529501d64d30af2aafc362d7f151 : SUCCESS in 10m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0617f4f070894fe2af18d8060a52ffbe : SUCCESS in 9m 27s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7b9cd1a9ec824d7985d50c1a321bc543 : SUCCESS in 1h 15m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ccbd308ce4064611b50b38f8bf6f21af : SUCCESS in 1h 14m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d21e3520e49b45d3b1258463f92935bb : SUCCESS in 1h 13m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dd19a5f29f714fada78f8697a8311273 : SUCCESS in 1h 27m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/42a1f7a8908e4a4aaf0f50d8c0187c2f : TIMED_OUT in 2h 05m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8090a699f8f44f7294091f35331b2d1b : SUCCESS in 57m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/59eb7145f67c4d6f908beab8ac4b22fb : FAILURE in 1h 04m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6e84e82fc0654fc4b975764fd9ca38e4 : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":6},{"id":"81ff4fd3866bebaee2d2be8f41b0e5a2a08f8188","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-11 19:27:01.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"065a7da6a37407597359574d4ba8e80038d33c6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-11 19:39:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16ebf050eb854e92a150fe303ed1ed8e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22d7bcfc91004c45931e760440a2fa78 : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fa84a1563d7549568ab171b87d45be8f : SUCCESS in 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a12b8bc0ef25e52bc438c18060f1a39cf2f5fd0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-11 21:04:22.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa0e49f7beb640a0808c01fe4e65dc7c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f93781c03b6e410ea11ab33003db3f51 : SUCCESS in 3m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/ec886b2eda4a42149918bf5ea16a4ae6 : SUCCESS in 1h 11m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b852c197b1d941f59644161a3ee5a1c1 : SUCCESS in 1h 22m 26s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8b3cb289a5f47eeae45caee15ee4a6c : SUCCESS in 1h 34m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12143dab3de94341b420a61830294ccf : SUCCESS in 1h 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45dde17147af4e49a683418331d6f1cc : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/247abacfeaa24d25bcd2150c9ed03dcc : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77c77be9e21b43ec9a806301b76d0a66 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35b51055a4b47a5b7a92c48190a10cd : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96fd0f74538249609e6c53c63cd67e84 : SUCCESS in 5m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a379486174a14a46878a098770fa963c : RETRY_LIMIT in 7m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9ff66849a23b4428ad1d95124ce4f249 : SUCCESS in 7m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e28413294d7d420ea73c16e18b281e09 : SUCCESS in 11m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d9cae34f190b4f7a97f70cb79b8836ef : SUCCESS in 1h 35m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b6dda7bbd6244ef79ee66653e6e17d5f : SUCCESS in 1h 33m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6c03326096624c6c92c28551d5101f4f : SUCCESS in 1h 14m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cff1030437d44602b4a7f69e640424e2 : SUCCESS in 1h 32m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/636c37827bef4bb697ca211c38a7b57e : SUCCESS in 1h 16m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b8a8aec3cc2412d84c3b2c432074163 : SUCCESS in 40m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c085806f3e474769bf4ea1bb0204cd6d : FAILURE in 55m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3ac6c88e583044db9dd7e2f433bf70aa : SUCCESS in 23m 25s","accounts_in_message":[],"_revision_number":6},{"id":"a69a91d03645da1d4c34311531238e224c6f3843","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-12 10:00:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d3efd40b7fee5e8157a48db345f4d8ffb777828b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-12 10:12:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d2d92bef08e43b384a4766e3e476a3a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d9df5d449294b07a621bece399a1cde : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/80488c0a5ee74f5984a9477f3cedd062 : SUCCESS in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a54bf1b8e8a5b719ccb0915b4b954099e4780b8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-12 12:15:32.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57b0d55f602c43168f5b61471104f21f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c3d56759f88d4dc59974c0612a8198bb : SUCCESS in 3m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/921512b880bd4ba797ab948c870ba2fa : SUCCESS in 59m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9dddc4940f5b4fed9671f54a295c0f0a : SUCCESS in 1h 09m 10s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cea7d273a2444894b75f05e638713e3e : TIMED_OUT in 2h 05m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bcbaffd57a042f1b021e90aeefa1ebc : SUCCESS in 1h 05m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26f611fbe4fc47149b4c4cf2d4d5693b : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb36d02d97f149f4a29c827e71eca1ec : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbc252662a3c4a6ba66da33014a8caf3 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17229bb79fdd46a58d878a2f87988edc : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59398be16abe43fc9dbe56534357ea9e : SUCCESS in 7m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/90a0378924d04c72aee9702bdfe53774 : RETRY_LIMIT in 6m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88e646b32f8b4c8fb90e86b37370ff54 : SUCCESS in 11m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6af16f57f85644aea86b1d0030695fd0 : SUCCESS in 9m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8ce51d74a4574d1690ac8730d6e3a72f : SUCCESS in 1h 20m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9710ce04ed2b4932a2a2783cd5a0fa5b : TIMED_OUT in 2h 05m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9425f46db8eb4ac8a7140af0d472c367 : TIMED_OUT in 2h 04m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/53502a25af9648889721c04210c5f755 : SUCCESS in 1h 35m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d37301ebaaf428cabee07dd2ccafe02 : SUCCESS in 1h 47m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67f22d989d034b7b86bd1a9c0a49f2b6 : SUCCESS in 1h 22m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c6703e12ab244950938e150f9f80808d : FAILURE in 1h 14m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f1e638cbf5b04ee38315c41efca2eb99 : TIMED_OUT in 1h 13m 30s","accounts_in_message":[],"_revision_number":6},{"id":"57fe89dd366b4d2b1d717c321da2447b15d1cfdb","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-13 05:15:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1f957e535719e3232c1b6cf7ec86de523474ab4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 05:21:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a994a59712b845b595b9ad4f5b64f639\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8ec0f32d04e41dea46eb45f7ec97642 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/175aaac09c8d418cb4c3d0d5c054847e : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"eb9aa6a8c3a07798e0fc0d1a9d545ad9f0ddf808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 06:53:58.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9737988dcd364989b63a0fe3b3b4f455\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ee6b04531eac4cf099a6ba3c7aa42ec4 : SUCCESS in 3m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/786a75db8f68471ca1176c134c8dcd63 : SUCCESS in 48m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/025d67cb3e4c4cd096aabe57f3ca0f5f : SUCCESS in 50m 48s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d589a25d99549b4a9e5ee9d230917da : SUCCESS in 1h 09m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed80031e7fe4410f9b9460134a96086e : SUCCESS in 48m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9da17bcd9abf40f4a799f1172b219018 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3094627723c4f949914781d15494095 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cdb8122445a4674bbaa83402ac0bab2 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58cd22dcd2394815b45c7f2e6c2de010 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a571e285cd045c592b40800cb0ce538 : SUCCESS in 5m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/59e0e08cff6b464da8271a0ce95bc3e2 : RETRY_LIMIT in 5m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a0086f0cde03469599ff0b0c9b8e3adf : SUCCESS in 8m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5b494f22682d4ae386feeabbd88ea44c : FAILURE in 8m 17s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/218dc6e4a59f4dd6adcdb26b0a169287 : SUCCESS in 1h 14m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2e602e9a4dfd4c308190a16229da8f63 : SUCCESS in 1h 16m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ec7df6a8342c4040b367e7c30b301762 : SUCCESS in 1h 21m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d9206b2d6e3d4f8594709b4b8b0e5fd6 : SUCCESS in 1h 31m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0a4426b37d114157a220c83175f3dab6 : FAILURE in 1h 20m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61053ea5c23c4b9fba740c3af17626c5 : SUCCESS in 55m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c1ff912bb37a4320b206997816dc6f6e : FAILURE in 1h 14m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ba7929a715f46d684dc6ade38acb7ae : SUCCESS in 43m 22s","accounts_in_message":[],"_revision_number":6},{"id":"d961e2602e6cf41e67cac272f27c8d957b4ce053","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-14 02:48:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"985eae8662a7d45a176bc9f764cd80e05efed06d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 03:02:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72973b7d55264c398e483af4c508563d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a2d34ad5d8c4c25b35f8ace4082aca8 : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b269036f82f94039bfc09c253e5324d0 : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"79508218dafe79ae58ecf90e69dc02ce2a7ce42b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 04:46:54.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00ad591c35ec4320a12eecb436381afb\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/450aee113701409dbe0c3c653c533fde : SUCCESS in 2m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/3fc8c88ce8d04048aca8cf4e9a739032 : SUCCESS in 47m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a8a72e6785d74227af951468d7032e79 : SUCCESS in 51m 51s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c0f06d7cb07f4bcdbbad10b73c5161ba : SUCCESS in 1h 18m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5c659f4c1524b8a96a7838aa1996ff9 : SUCCESS in 51m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9bc78bcfad94e62b7c01324f9032c6b : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/037735d0fe9547a29fdeb9c00c80343e : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07df3e9ddae74c04b9395bc0e71112b7 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00fb79bc778140cc943425539230d19a : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa650c5e132942c6ad346922045a51a6 : SUCCESS in 5m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8b5026a919c34ce891435c28671c37f6 : SUCCESS in 12m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7039ae1469f14472a3968c90cf1909a8 : SUCCESS in 8m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/52b0e74513fe4214a6df8f61476520fe : SUCCESS in 7m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f047ef2b515b4c74821cfb45bc8cd4d7 : SUCCESS in 1h 25m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/678181837f1e4cdc9879320efd1496b0 : SUCCESS in 1h 21m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1811dc217d9742f8935d65fac44004c5 : SUCCESS in 1h 18m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9539deb1bd5d413a8f408721d4ec19ec : SUCCESS in 1h 36m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ea2fc1a867f84bfcbf519a727ad1e6f3 : FAILURE in 1h 40m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1edcbf9587e4239b1649aaf389f2ef1 : SUCCESS in 53m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e7dd084a94cc44b4a0553889d8dce97f : SUCCESS in 1h 10m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ee0126b17db4db0943657a1f6229aee : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":6},{"id":"6a5720b1c8bda2a1eee6b05058a73a8cd8fbdc7c","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-14 05:36:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"39fa52483143219bc2d03bc1c835b9802e3007bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 05:48:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dd612d53ea24c289fb1f55ebdfd865a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7327e719d7a4d9588e48142721214dc : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62d34e158630407cbe4399caf9c78c7f : SUCCESS in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3c3c47df41f9d49b7c3c6bd5a3dc20f8a0550857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 07:46:34.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e65f5d21c49451084ce7d58b0bf4880\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/02b32ac0a4f64754879a3fba0d1a872f : SUCCESS in 2m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/e1357279bed442249155114f162ec532 : SUCCESS in 1h 44m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/75b770ae2847426daf4e477966a33dee : SUCCESS in 1h 44m 02s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/039f6ae7ef97488cbb844fe6dc1846fd : SUCCESS in 1h 25m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42b5748cac354053828901dbb0d69ffd : SUCCESS in 1h 05m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99820c2679a046859f2f906a8fea51a4 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fbc44f7f9d644c8b6892ac3e5db5865 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/365bd17195bb40e9a1efedc44d5a8661 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e2aab61a9fd4de2be996e1d2d41c9a2 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d68633366d6f4c4b9aaf22ba7fb3e07d : SUCCESS in 4m 42s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3c2b46787c264209bf9d350b0c1613a8 : SUCCESS in 16m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d5731cf1a90e4f09ad68f6e27ce12778 : SUCCESS in 9m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c10f0755d54948f48e205138d721b662 : SUCCESS in 8m 26s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c7af236e902e407b8356a425fdcae7c7 : SUCCESS in 1h 20m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8dfed5d5914d4d8b8096610e784cd31d : TIMED_OUT in 2h 03m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6f790f1240554d8abeb50c88b487dee0 : SUCCESS in 1h 18m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4c7ebba987e74f3a83144d209f2f7591 : SUCCESS in 1h 20m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/954e3fee9ef74a958a4b42cd332581bc : SUCCESS in 1h 48m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/517563e5b3d84ba081b23829ea38a8ad : SUCCESS in 45m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d4291992fe9943929b48800450b027fc : SUCCESS in 1h 58m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3e35d0317c714258acee4836b8dd72aa : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":6},{"id":"9f30d77635829891196a285aa3475b6e304777e8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-14 12:59:20.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9d3c84ab29fb506f6dc83a9cfdccca20f670e8c1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-15 05:43:43.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a3376a295ff46db5739d266c64c2237d069c8bf4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 06:00:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eb79fc5caba4090b314afced51b1c5d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2b9a956974a4e6dbdab7af915bad9c1 : SUCCESS in 13m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/46c705c8f2e84086942b986e2e5c463d : SUCCESS in 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"139894fc639824bad1f095864f64843011fe8e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:55:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcb64c25dd384e50bc34fbe3011da772\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/28dfcedbd36b4946bbd5e8b39cef3ae3 : SUCCESS in 2m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/094a529148994a06a6a118cdab4a5539 : SUCCESS in 1h 04m 12s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/07529037e4744544ab6e38c147c58313 : SUCCESS in 1h 15m 51s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c8ae5cb5ec24e44830c71565d8473c1 : SUCCESS in 1h 35m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a262a7f981d343488cb3326af8655f02 : SUCCESS in 53m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b27e7501d444fa1834e88ad26d5ac94 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8796882bf54048228eff0dd40b15d5b8 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d840443480d458f841875de2c554ce3 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cad3d66d0e434d8ebe2f0a0495bbb0d9 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35ece34452914eb8b6a9fea3cd013315 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/122b2d2902254852ae20b02478ce8ba4 : SUCCESS in 15m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b864c5d0bd94675a739e0fd6c922fb8 : SUCCESS in 9m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/aa41891d8fc74b7ba45cfe3636ae2dc8 : SUCCESS in 8m 57s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ec9830caf2674b6e8ab1618396f31b3b : FAILURE in 1h 16m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/338c9487ccf14d9681eb489d45cc962c : SUCCESS in 1h 20m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9e3e545a48b14d46baf3dcbf65615b02 : SUCCESS in 1h 35m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4e70e0496048494c9ed1675e73f8e75b : SUCCESS in 1h 23m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b1555c0f0d5d47229e7cba153be500a1 : SUCCESS in 1h 49m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aab8d109ae2b4d1cab576d9c3b925810 : SUCCESS in 42m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8bc81718e2a6417ea0c1253b49b9add9 : SUCCESS in 1h 18m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/33629d464eb742f7bfbee527a8ecade9 : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":6},{"id":"f7f5f437a5197390f75b6635236069af4624c2b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-15 18:10:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"e7994ab90d117933db831ebcee661bf93349e6d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:18:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/562cf64717034a0d8cc6df00d9783b4a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45924502a96a41f594cc375d4fe642cb : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec21bb2c52f4489186125cb2d160cab8 : SUCCESS in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ce9ec337a9fbfe295aef61456e79814a9e5a145e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 20:28:25.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/797600518f4d4736a258ac4ba03b5270\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac5e3ee920c541c28bd1674ef8c66369 : SUCCESS in 2m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/5017de340d9f4d48af30ef9fc524469f : SUCCESS in 1h 12m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2d409bec7a19434cb70b69a336562956 : SUCCESS in 1h 10m 34s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04c4bd017aa842e2a0f11e819a317ea4 : SUCCESS in 1h 49m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc8f812be5af4d9794774570b4455415 : SUCCESS in 1h 00m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3be1681e0b864a10bbd9f9358fa9ebed : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e83f45a6e3f54bd5a84a567a9a610e11 : SUCCESS in 6m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/206fac6418374f0cb54dfcb6d2154fd2 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74b3135eacf42e9b052f70d6c78f7b6 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a62f95fa21ac4ff787b7768b68317e65 : SUCCESS in 5m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/14c65ec3029c48a89ee69a621a78fe2e : SUCCESS in 14m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/05fc8e5bd73c4a95b5fb252a1e879f7b : SUCCESS in 7m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8b655bcd61e84afaabb0b6f4762b89cc : SUCCESS in 9m 13s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6e7335df75c3465cb78f4ad9923ae884 : SUCCESS in 1h 16m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7d9f8ecbaf944f2098051976f7a9dd09 : SUCCESS in 1h 21m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0f928932d0fa43debfc12c9520163f59 : SUCCESS in 1h 23m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9e8858968a2c4898aa48a9cdad651672 : SUCCESS in 1h 13m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d8c23db1e54e4477872e4913aea435e2 : FAILURE in 1h 18m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd2646f64c7f4e568c51d2d8d2c1eec9 : SUCCESS in 42m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/97f7af6a5e2647378399737cc58c19a6 : FAILURE in 1h 33m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c4bbde531f5d40fd9216110bce85fb43 : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":7},{"id":"42f7d7c24d9db61c819c071da9e6b7c1a2609e75","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-16 02:12:17.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b7e7cdc5322fdcc25cc5bce2c9b0e9db412ab21f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:31:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d010373d9a047f88cee906b1b1d3389\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c81fd25d67ed4a3186779efd816eba41 : SUCCESS in 13m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7990623740a24fd092cdd52819a70db2 : SUCCESS in 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"00b79fac8109797182356b5ee0ac1884973e5f9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 04:46:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/545c45436b04435b80f265b8b99d7798\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e02316ef09b44955b71a90b83fa964f4 : SUCCESS in 3m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/3f946761896942de841e2e5172fbd200 : SUCCESS in 57m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/055e2a2df07749f499104a27b8ce04cb : SUCCESS in 1h 00m 11s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a8654802ce141c5b17d321124f062f3 : SUCCESS in 1h 18m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45833ce1c76f42a5866aa9972ab21d0b : SUCCESS in 1h 08m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44639b2d0ab847faac713f952e61229a : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4ffbbea769245e381bfbf6073bbdaa8 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ab6706ce51245ad97e501d14ff2dea9 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a0f145a7564e0b98f3f97ae5e7e1b5 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05d1cb6ba4df4d09a4c25a411d95dfe9 : SUCCESS in 5m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b3b244c8d23a4021805192891249b914 : SUCCESS in 13m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e3ebc5c9167c49adb878b106b910d76e : SUCCESS in 8m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1942990ebba942188e87990d36861717 : SUCCESS in 7m 48s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/923ac03032324068b34c28ad65786c6a : SUCCESS in 1h 10m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5cac6667a3a34dc894c059d7d8362e63 : SUCCESS in 1h 25m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e191f514ebbd4ccd923b554310723cec : SUCCESS in 1h 23m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/117a80f79a214160b0274248d6dcc06c : SUCCESS in 1h 32m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9d90fbc7e8354f8a92e6310746f2e7c3 : SUCCESS in 1h 30m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dca8d61e5fd4c62b830da0b277b8b1b : SUCCESS in 54m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f0d6ac253186448fa2ef21f52a76b92f : SUCCESS in 1h 09m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/236efab3df1b43228a62416bb6a5c517 : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":7},{"id":"f8a217c0ba3e86630b23c9213ba504d4618a8f21","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-16 06:45:36.000000000","message":"Patch Set 7: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0f23d37f8882b32d94974690a8903b788da5336d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-16 11:39:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"69f0893a8b92d863f0a71a3a027c74b67b945a47","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-16 11:43:31.000000000","message":"Patch Set 8:\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e4be5b439aebba1f0e2d40c4ec80eeb992706771","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 12:02:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bdb604b39da4ff3bd8762a9f4503b93\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a30cfd4ba0c0452395f1766d5e29ba36 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/483a52de5ab64178813260a019f9b874 : SUCCESS in 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fb63cb1e4ed86dcc4f1519eac69c2d0233fa47d8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-16 13:52:58.000000000","message":"Patch Set 8: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"fa40c90906c3fa526f24d152e839f1a01589c38e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 13:54:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1f9bfe854da4607b66a6ee0eca8fcd0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/804217a1e8a445e0874ea91f3e6bbf86 : SUCCESS in 3m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/9922d9f9bfb94bbe9f542cd7b3817578 : SUCCESS in 48m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4137c6ba3cdd4678ae0ffa932db91cb5 : SUCCESS in 52m 53s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e63b6486bb3c44da88e4fab360ecbd52 : SUCCESS in 1h 08m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14c6899937a942dd9e7e348d99d0397e : SUCCESS in 54m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/401c261d3f59447b91d8d060abcda4c2 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb4ee11353cd474ba38d69aa970e1c76 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dfe8046a8464f648306a858f50e2b0e : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf28206ea3c4f8a9054263504951269 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10c3a4d81beb4034a2653742a984d1aa : SUCCESS in 7m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7668b805b1ca4b63a40af4333ea1ab5f : SUCCESS in 12m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1d574af6559142c08124c991e21fc0ef : SUCCESS in 9m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/19421ac7b4d340bca7a126c2a98129e6 : SUCCESS in 11m 41s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c397c4307ea94727901b26fbd097b446 : TIMED_OUT in 2h 04m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1674b5cd5bec415b99c1c72a6e45968c : SUCCESS in 1h 17m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/53eff6909d034320bf7a81ca58015f5b : SUCCESS in 1h 17m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/56e1f191b131421294325dc0e2bc00cd : SUCCESS in 1h 24m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a73b6e167b8e49a39a13cad1f6a15129 : SUCCESS in 1h 44m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a1012fba6bd478faf491999151eb4b9 : SUCCESS in 51m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b74dcd8bcfd4e328eb9da9174fa446d : SUCCESS in 1h 19m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/08a0d3393076468dbb0f197c0cb772de : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":8},{"id":"28beb3c433b7cd217230ccc8aa0bdbc323331d84","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-02-16 14:39:33.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7fbaa970934df1c64b9b2db15b93fc1846ef924a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-16 18:27:40.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5d3f2ce4ac7d1f21861cb8806220f02925355d97","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-20 13:40:43.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"20ab2387637359fdc2520a5da7b5f24a926a5f42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 13:41:53.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"49d726585fd7b57534ee5eadbd552126d34d9ac6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:13:36.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/242461792e1d4367b83ac51802d73069\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/31e62562025f4efb930dba6ba8a252c4 : SUCCESS in 4m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/9d55cd5b2a2449858def70be7bba87a2 : SUCCESS in 50m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/57186c2c1ec64b2eb32920057ac10410 : SUCCESS in 1h 28m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f37fb82fabf44481b6b0c6ce7beaebaf : SUCCESS in 57m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2053af20b680421dab19c8537c39eed0 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/660476fcba0542c9b2ea8ac8404e9c75 : SUCCESS in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75e027a818bc435aa8f8af5f6899eb4e : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47ca73e2098542da8b3a8f0962b8c28b : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03f2aca3055146719db9f08d30ed41de : SUCCESS in 5m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3bf0a593c9134e5aa5c93265bafad9f6 : SUCCESS in 9m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e9092339a21e4cbfbdb31c403b1387a9 : SUCCESS in 10m 12s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/97cff81c8fdf4c8dbed696bb7d122b6e : SUCCESS in 1h 19m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a2be9b9200c14f3ba97503c1d6121678 : SUCCESS in 1h 13m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5135101c4d444cb2b62e6f4492c42e7f : SUCCESS in 1h 24m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cac032c688994475a819c8b444eb040f : SUCCESS in 53m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bd7e423bc504747b0d3912c4c22f653 : SUCCESS in 52m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/82c87f712d834ab9ba0faa34005dce5a : SUCCESS in 1h 06m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a3cf7defb1af4851a442f21dd7b07674 : SUCCESS in 30m 55s","accounts_in_message":[],"_revision_number":8},{"id":"e97513fea35c79a8b075958e560dcbabd64aba9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:13:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"1d2a534fe09e1d9e7f11556c86c9e8dc6c71b64f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:16:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96828240c6074982ab9e6691df59b715\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e73d6b767e4e4c29841e6fa7e4e25c75 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddab0ca9614a410ea9016f2b60542d74 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8c04d19e8858594fe687f555c9be99ce6b16dc16","revisions":{"ba03500a0679276535f681e3ac94b51eefc88475":{"kind":"REWORK","_number":1,"created":"2023-02-02 10:42:47.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/22/872522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/872522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/872522/1"}}},"commit":{"parents":[{"commit":"907c56265438da43c13bf1a3369d5459b1267e34","subject":"Merge \"Enforce image safety during image_conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/907c56265438da43c13bf1a3369d5459b1267e34"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-02 10:24:00.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-02 10:24:00.000000000","tz":0},"subject":"Enabled new defaults and scope checks by default","message":"Enabled new defaults and scope checks by default\n\nEnabling the enforce scope and new defaults by default in glance\n\nChange-Id: I0808dc0b1b34b527e38aa137c1dd25e1fc06409f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba03500a0679276535f681e3ac94b51eefc88475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba03500a0679276535f681e3ac94b51eefc88475"}]},"branch":"refs/heads/master"},"bb72fc411c5418fbbcb4a548d6a2ca0e3af30cbf":{"kind":"REWORK","_number":2,"created":"2023-02-08 08:00:44.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/22/872522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/872522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/872522/2"}}},"commit":{"parents":[{"commit":"3929db2c730f7d977e72c713bff02be725219b9c","subject":"Merge \"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3929db2c730f7d977e72c713bff02be725219b9c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-02 10:24:00.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-08 07:40:15.000000000","tz":0},"subject":"Enabled new defaults and scope checks by default","message":"Enabled new defaults and scope checks by default\n\nEnabling the enforce scope and new defaults by default in glance\n\nChange-Id: I0808dc0b1b34b527e38aa137c1dd25e1fc06409f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb72fc411c5418fbbcb4a548d6a2ca0e3af30cbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb72fc411c5418fbbcb4a548d6a2ca0e3af30cbf"}]},"branch":"refs/heads/master"},"77acc68ce4bb41ec9e193a03be43ba2894ae662c":{"kind":"REWORK","_number":3,"created":"2023-02-08 09:16:58.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/22/872522/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/872522/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/872522/3"}}},"commit":{"parents":[{"commit":"3929db2c730f7d977e72c713bff02be725219b9c","subject":"Merge \"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3929db2c730f7d977e72c713bff02be725219b9c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-02 10:24:00.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-08 09:16:58.000000000","tz":0},"subject":"Enabled new defaults and scope checks by default","message":"Enabled new defaults and scope checks by default\n\nEnabling the enforce scope and new defaults by default in glance\n\nChange-Id: I0808dc0b1b34b527e38aa137c1dd25e1fc06409f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77acc68ce4bb41ec9e193a03be43ba2894ae662c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77acc68ce4bb41ec9e193a03be43ba2894ae662c"}]},"branch":"refs/heads/master"},"ec9da06797025d8ee8d75fb44dbc12d6abd2e02e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-10 14:05:53.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/22/872522/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/872522/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/872522/4"}}},"commit":{"parents":[{"commit":"b27b26a6a8a315fe1c0e28fd0240f26488081409","subject":"doc/useful-image-properties: add missing `` to close a code block","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b27b26a6a8a315fe1c0e28fd0240f26488081409"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-02 10:24:00.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-10 13:53:08.000000000","tz":0},"subject":"Enabled new defaults and scope checks by default","message":"Enabled new defaults and scope checks by default\n\nEnabling the enforce scope and new defaults by default in glance\n\nChange-Id: I0808dc0b1b34b527e38aa137c1dd25e1fc06409f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ec9da06797025d8ee8d75fb44dbc12d6abd2e02e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ec9da06797025d8ee8d75fb44dbc12d6abd2e02e"}]},"branch":"refs/heads/master"},"fb335a97e414df50d74bf4aeedfc6750b788143e":{"kind":"REWORK","_number":5,"created":"2023-02-11 11:47:41.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/22/872522/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/872522/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/872522/5"}}},"commit":{"parents":[{"commit":"b27b26a6a8a315fe1c0e28fd0240f26488081409","subject":"doc/useful-image-properties: add missing `` to close a code block","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b27b26a6a8a315fe1c0e28fd0240f26488081409"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-02 10:24:00.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-11 11:11:07.000000000","tz":0},"subject":"Enabled new defaults and scope checks by default","message":"Enabled new defaults and scope checks by default\n\nEnabling the enforce scope and new defaults by default in glance\n\nChange-Id: I0808dc0b1b34b527e38aa137c1dd25e1fc06409f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb335a97e414df50d74bf4aeedfc6750b788143e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb335a97e414df50d74bf4aeedfc6750b788143e"}]},"branch":"refs/heads/master"},"d2e13cbdc8e6268b37f8db9cbd790a2803d0bf5e":{"kind":"REWORK","_number":6,"created":"2023-02-11 11:55:49.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/22/872522/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/872522/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/872522/6"}}},"commit":{"parents":[{"commit":"b27b26a6a8a315fe1c0e28fd0240f26488081409","subject":"doc/useful-image-properties: add missing `` to close a code block","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b27b26a6a8a315fe1c0e28fd0240f26488081409"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-02 10:24:00.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-11 11:41:40.000000000","tz":0},"subject":"Enabled new defaults and scope checks by default","message":"Enabled new defaults and scope checks by default\n\nEnabling the enforce scope and new defaults by default in glance\n\nChange-Id: I0808dc0b1b34b527e38aa137c1dd25e1fc06409f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2e13cbdc8e6268b37f8db9cbd790a2803d0bf5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2e13cbdc8e6268b37f8db9cbd790a2803d0bf5e"}]},"branch":"refs/heads/master"},"6f54b6ba9f1420ec4ef9119fc381e81973171090":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-02-15 18:10:36.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/22/872522/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/872522/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/872522/7"}}},"commit":{"parents":[{"commit":"ce8251f8e1656105834456db1adb1deaeced3903","subject":"Merge \"Drop tag assertion from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ce8251f8e1656105834456db1adb1deaeced3903"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-02 10:24:00.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-15 17:03:50.000000000","tz":0},"subject":"Enabled new defaults and scope checks by default","message":"Enabled new defaults and scope checks by default\n\nEnabling the enforce scope and new defaults by default in glance\n\nChange-Id: I0808dc0b1b34b527e38aa137c1dd25e1fc06409f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6f54b6ba9f1420ec4ef9119fc381e81973171090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6f54b6ba9f1420ec4ef9119fc381e81973171090"}]},"branch":"refs/heads/master"},"8c04d19e8858594fe687f555c9be99ce6b16dc16":{"kind":"REWORK","_number":8,"created":"2023-02-16 11:39:09.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/22/872522/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/872522/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/872522/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/872522/8"}}},"commit":{"parents":[{"commit":"81da9ccbdee88e77f06e42c4f121b68cf2c1389e","subject":"Merge \"Further robustification of format_inspector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/81da9ccbdee88e77f06e42c4f121b68cf2c1389e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-02 10:24:00.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-16 11:11:31.000000000","tz":0},"subject":"Enabled new defaults and scope checks by default","message":"Enabled new defaults and scope checks by default\n\nEnabling the enforce scope and new defaults by default in glance\n\nRelated blueprint secure-rbac\n\nChange-Id: I0808dc0b1b34b527e38aa137c1dd25e1fc06409f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c04d19e8858594fe687f555c9be99ce6b16dc16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c04d19e8858594fe687f555c9be99ce6b16dc16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
