)]}'
{"id":"openstack%2Fglance~872990","triplet_id":"openstack%2Fglance~stable%2Fzed~I3ec5a33df20e1cfb6673f4ff1c7c91aacd065532","project":"openstack/glance","branch":"stable/zed","topic":"fix_format_inspector_vmdk-stable/zed","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-03-09 14:12:41.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-03-10 09:53:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-02-09 14:37:34.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I3ec5a33df20e1cfb6673f4ff1c7c91aacd065532","subject":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX","status":"MERGED","created":"2023-02-07 16:49:15.000000000","updated":"2023-03-10 12:53:20.000000000","submitted":"2023-03-10 12:52:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872990-fix_format_inspector_vmdk-stable/zed","cherry_pick_of_change":871831,"cherry_pick_of_patch_set":4,"meta_rev_id":"453d7348e781c9ce0be203903af5a6aba2abc6c6","_number":872990,"virtual_id_number":872990,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-10 12:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2023-02-21 06:16:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2023-03-09 14:12:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-02-09 14:09:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},{"value":2,"date":"2023-02-09 14:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2023-03-09 14:12:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-07 16:49:15.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"state":"CC"},{"updated":"2023-02-07 18:35:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-08 05:35:34.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-02-08 05:35:34.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-02-09 14:09:31.000000000","updated_by":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"reviewer":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"state":"REVIEWER"},{"updated":"2023-02-21 06:16:30.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"72f2124f090cdcf4c5178b128304367927dc34c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-07 16:49:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"016d39e6375f64a3a27bbeae8349d996f150ee31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 18:35:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ef218be45a246378e7d65cc251ae7d2\n\n- grenade https://zuul.opendev.org/t/openstack/build/d0c28a0625da4369ad2b94b37e59431c : SUCCESS in 1h 00m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3265b5ea69214d3e978ca2a99573d84f : SUCCESS in 1h 38m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f503eadde6324185898080e63c01b5bd : SUCCESS in 58m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b6832cce44a4e50852260e77e104347 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ff7911539b34f579a7f5c74d3e517d7 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89b532d1a407439d9cd411384c8802ba : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46c76bf391f245cd9bddc441ce87e28e : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c3203c8ffed4f46bddf4f956d30433e : SUCCESS in 6m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/650a2c00045841608f0091d2350f806a : SUCCESS in 12m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/455e04a5ade9479f83711a409bbce80b : SUCCESS in 9m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2fe75df212e943c6adb97a1933d588c8 : SUCCESS in 12m 06s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/95bafba22aec4c7e8a1e5dba6095fb69 : SUCCESS in 1h 25m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a040c89abb614ae9966d772ab1e96f58 : SUCCESS in 1h 30m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0756bdb675784c21ae2f76e5e3a2c0ba : SUCCESS in 1h 20m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dc804aa2a8c74b2bbf4da3015e8d48df : SUCCESS in 1h 26m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdca594cadaf420dbe5b1d0764219b1b : SUCCESS in 1h 09m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dc2a2d36f5d24df39f1c55e8bba25440 : SUCCESS in 1h 06m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c64e086c74b14073865291797cd8c03a : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":1},{"id":"82988a5498cb6e2eec70a484c748511abeb235db","author":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"date":"2023-02-09 14:09:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f6c567f830899ff2df24322140292672a857fa8c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-09 14:37:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0d43947822fd7d01567a3f2700463b617e145a8","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-21 06:16:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6594f86b14b5e195e2e77e5aee0922a680dadce2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-03-09 14:12:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8b09a2d57fd3d26dd19ea7e9506dddf9221a9e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 14:13:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"726097b9e298dd1d2e47c07a0e153ef81a8fc24d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 15:52:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66fed944ec39466d86a067d118f752c6\n\n- grenade https://zuul.opendev.org/t/openstack/build/86fc90e066144766bd7dcc3cbf198721 : SUCCESS in 55m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f0e8d11f47ab4444bb7ade8e4d298e1b : SUCCESS in 1h 25m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6945bdc4a8d448a6b3286ca3e293f082 : SUCCESS in 54m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/596b92a0a6314db192f0d06353e056ec : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e300eaeff66348cf828eddfefe3f5c7c : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d407277e573a4c74a291658f05c8c18a : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f244554db01f4e758ec262cb560d48de : SUCCESS in 6m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45eb53eae176403ebc3c6ae2c271f37e : SUCCESS in 7m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/67310b8ddb8b49bf82a859dbd7b5288f : SUCCESS in 8m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5f41c05854dd47b9b519b028fdc41726 : SUCCESS in 1h 08m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fb6783101b1e46f7a12bb31014360578 : FAILURE in 1h 23m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bacbc035a774f66b8193da6c9978bea : SUCCESS in 40m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7fce4c3a56254f5787285a1887c9b526 : SUCCESS in 59m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f3edf8fb9fa44f5eb198d73ba756a63b : SUCCESS in 25m 38s","accounts_in_message":[],"_revision_number":1},{"id":"01f3a668715a017f80b230c099ee9ea07747b4aa","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-03-10 09:53:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14f65ffa550c255f66cf75ad8b2b5c32755c1e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 11:29:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d1185ab35304a309a39a227113b42a8\n\n- grenade https://zuul.opendev.org/t/openstack/build/919c17ea2f334d649422a9516d1422ae : SUCCESS in 54m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/62dfa23bb4a0414c91fc5ebfe9fe7829 : SUCCESS in 1h 17m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a07db14b9eeb45439b79edd19a61c059 : SUCCESS in 47m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f8c517e2d0c45bf82b8104e9f25e9e3 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a8dc2f788d145cfabd2fd37b98b22d1 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ea76b6df4b04b29981eba3ff9b27ffe : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9813ce08c24b48fa93ec772a97d3609d : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc07baf819f45aea87de749ea1e4408 : SUCCESS in 5m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8da037a21dce4233af5a08a427b42125 : SUCCESS in 12m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/28e249486bc04c568bc2a5f13dbeb4c8 : SUCCESS in 8m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c0b2162c0db245feac61f7b859066726 : SUCCESS in 7m 41s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/86823c3b185b43cb8c68d2501c7b7616 : SUCCESS in 1h 12m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2940910ce07248219e19a43b5e4952e1 : SUCCESS in 1h 12m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0db669ec9d574f9dbc50c0455ea8d0af : SUCCESS in 1h 00m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1d9a3ecb50bf490faaf68b0cabf68592 : SUCCESS in 1h 28m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0864344e06d44f1c89a6f954af9917e2 : SUCCESS in 39m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0ed1578824fd49d8aa096a790f6549c1 : SUCCESS in 1h 04m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ac51529e53f4563b3e2502b004547cb : SUCCESS in 31m 24s","accounts_in_message":[],"_revision_number":1},{"id":"df2f5eb7b9f79a2ed0c03cf305889f9479e49234","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 11:29:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8891e3ba2166722633298e8a772d8a1d51b58bba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 12:52:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0adc4486090e41c39ecf02100f21c037\n\n- grenade https://zuul.opendev.org/t/openstack/build/e7f1a7ac789e4cc1beb875612869c82c : SUCCESS in 1h 00m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8beedcbd744841949aa1b5b216fdb276 : SUCCESS in 1h 17m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa715b980d6f436d859277e0d1719263 : SUCCESS in 56m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d50f3c179704930b3aa9bbcf00aa042 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/280d5ea7561644468d35a7952ef50768 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69ce0c5cdeb745cd9b1f4e7313a9cb5e : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507b70e046bf43f48d0c92b23c5e43c9 : SUCCESS in 6m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cbad564751a347e2ae9166dab15ec34d : SUCCESS in 7m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fe57d5a674414d0593d1d2047cabadbd : SUCCESS in 7m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1b6b265992364321bde48f98c711fbbc : SUCCESS in 1h 16m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/56683d557e2647e389f216b7abe59d37 : SUCCESS in 1h 14m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/987e3e7541bc4796ae3004f318cc250b : SUCCESS in 45m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6e71573d80ab43d286e3e25740c7b636 : SUCCESS in 1h 10m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0ab94a713508484e801f3af862ef1195 : SUCCESS in 27m 57s","accounts_in_message":[],"_revision_number":1},{"id":"73be9eeb5eb3a2d9d84edc233a57dfd247a465dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 12:52:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"453d7348e781c9ce0be203903af5a6aba2abc6c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 12:53:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98066931462645b3bb0f3e327815dfd0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c56d88ebaf147fe882b4a148c4d347c : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06a18202ab52c64803f044b8f848ed1c160905d2","revisions":{"06a18202ab52c64803f044b8f848ed1c160905d2":{"kind":"REWORK","_number":1,"created":"2023-02-07 16:49:15.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/872990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/872990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/872990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/872990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/872990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/872990/1"}}},"commit":{"parents":[{"commit":"b1e5292248fefb7cd1ff4292c9b94d336f4cf73b","subject":"Merge \"Enforce image safety during image_conversion\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1e5292248fefb7cd1ff4292c9b94d336f4cf73b"}]}],"author":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-01-25 10:53:09.000000000","tz":60},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-02-07 16:49:15.000000000","tz":0},"subject":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX","message":"Limit CaptureRegion sizes in format_inspector for VMDK and VHDX\n\nVMDK:\nWhen parsing a VMDK file to calculate its size, the format_inspector\ndetermines the location of the Descriptor section by reading two\nuint64 from the headers of the file and uses them to create the\ndescriptor CaptureRegion.\n\nIt would be possible to craft a VMDK file that commands the\nformat_inspector to create a very big CaptureRegion, thus exhausting\nresources on the glance-api process.\n\nThis patch binds the beginning of the descriptor to 0x200 and limits\nthe size of the CaptureRegion to 1MB, similar to how the VMDK\ndescriptor is parsed by qemu.\n\nVHDX:\nIt is a bit more involved, but similar: when looking for the\nVIRTUAL_DISK_SIZE metadata, the format_inspector was creating an\nunbounded CaptureRegion.\n\nIn the same way as it seems to be done in Qemu, we now limit the upper\nbound of this CaptureRegion.\n\nCloses-Bug: #2006490\nChange-Id: I3ec5a33df20e1cfb6673f4ff1c7c91aacd065532\n(cherry picked from commit d4d33ee30f303f783c0640cd72acb31b313e1164)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06a18202ab52c64803f044b8f848ed1c160905d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06a18202ab52c64803f044b8f848ed1c160905d2"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
