)]}'
{"id":"openstack%2Fglance~873451","triplet_id":"openstack%2Fglance~stable%2Fzed~Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f","project":"openstack/glance","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"35025":{"account":{"_account_id":35025,"deleted":true},"last_update":"2023-02-17 06:42:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f","subject":"Update the RADOS authentication link","status":"ABANDONED","created":"2023-02-13 09:51:32.000000000","updated":"2023-02-17 06:42:18.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":865227,"cherry_pick_of_patch_set":2,"meta_rev_id":"0ff1c575616d3577cdd49ae7cc39954ed0df045b","_number":873451,"virtual_id_number":873451,"owner":{"_account_id":35025,"deleted":true},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-13 11:00:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":-1,"date":"2023-02-15 05:49:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 11:00:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 05:49:18.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"f7e140b46fd883db91799e910b8b8506ddadd072","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2023-02-13 09:51:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2d4ce501148b99ffddbadbc905c887e58e92e606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:00:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a3e2421b52649f0a0473093e3e763f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c6aa0e4bc214f0180122fd507688ade : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb8c51c482a4b73af91a4bd7e142be3 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0ad9c12e6683d6d3bd12fbabd597a1ab9aeaac96","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-15 05:49:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ff1c575616d3577cdd49ae7cc39954ed0df045b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35025,"deleted":true},"date":"2023-02-17 06:42:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc2f571e970a2c95e0d7916652b97bf401581f69","revisions":{"fc2f571e970a2c95e0d7916652b97bf401581f69":{"kind":"REWORK","_number":1,"created":"2023-02-13 09:51:32.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/51/873451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/873451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/873451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/873451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/873451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/873451/1"}}},"commit":{"parents":[{"commit":"b1e5292248fefb7cd1ff4292c9b94d336f4cf73b","subject":"Merge \"Enforce image safety during image_conversion\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1e5292248fefb7cd1ff4292c9b94d336f4cf73b"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-11-22 05:46:03.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-02-13 09:51:32.000000000","tz":0},"subject":"Update the RADOS authentication link","message":"Update the RADOS authentication link\n\nThe original link has failed, update the link\nhttps://docs.ceph.com/en/pacific/rados/configuration/auth-config-ref/.\n\nChange-Id: Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f\n(cherry picked from commit 3f9d58dedd5e5f5cd5034abbc86778dcf5d4c883)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc2f571e970a2c95e0d7916652b97bf401581f69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc2f571e970a2c95e0d7916652b97bf401581f69"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
