)]}'
{"id":"openstack%2Fglance~873454","triplet_id":"openstack%2Fglance~stable%2Fwallaby~Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f","project":"openstack/glance","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"35025":{"account":{"_account_id":35025,"deleted":true},"last_update":"2023-02-13 14:13:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f","subject":"Update the RADOS authentication link","status":"ABANDONED","created":"2023-02-13 09:52:07.000000000","updated":"2023-02-15 05:49:50.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":873453,"cherry_pick_of_patch_set":1,"meta_rev_id":"3cb5a69eaa20031a2387a6d2b4a8e09acbe4cd99","_number":873454,"virtual_id_number":873454,"owner":{"_account_id":35025,"deleted":true},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-13 11:05:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 11:05:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 05:49:50.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"}],"messages":[{"id":"de6e414f5feaa1d151aa4d6ec49adfeb3662f271","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2023-02-13 09:52:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"13986a5a009d23653cd9337cb21ef1118549279b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:05:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aff6c08461f84b38ad9f577d1c0b6d21\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e9c646b613466db7adc53380e9f5aa : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0cabcd9f37b4d7a8f691ed3ecd5050d : FAILURE in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cdb84947f7b67498574b493b0e878c270c0b21ad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-13 14:13:12.000000000","message":"Abandoned\n\nThis branch is too old, we will not be merging this.","accounts_in_message":[],"_revision_number":1},{"id":"b436520bc37899abf60676b7e0cb2c041f2ff82c","tag":"autogenerated:gerrit:restore","author":{"_account_id":35025,"deleted":true},"date":"2023-02-14 01:56:27.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"f2e945b77d83b95b23f5309753ba2255099d8fa2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35025,"deleted":true},"date":"2023-02-14 01:59:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"3cb5a69eaa20031a2387a6d2b4a8e09acbe4cd99","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-15 05:49:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d62a78274d09c49d664d72a477af35a879ab1581","revisions":{"d62a78274d09c49d664d72a477af35a879ab1581":{"kind":"REWORK","_number":1,"created":"2023-02-13 09:52:07.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/54/873454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/873454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/873454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/873454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/873454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/873454/1"}}},"commit":{"parents":[{"commit":"c555fef175236affd1d5f8776a56044bddb40600","subject":"[stable-only] Remove glance-code-constants-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c555fef175236affd1d5f8776a56044bddb40600"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-11-22 05:46:03.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-02-13 09:52:07.000000000","tz":0},"subject":"Update the RADOS authentication link","message":"Update the RADOS authentication link\n\nThe original link has failed, update the link\nhttps://docs.ceph.com/en/pacific/rados/configuration/auth-config-ref/.\n\nChange-Id: Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f\n(cherry picked from commit 3f9d58dedd5e5f5cd5034abbc86778dcf5d4c883)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d62a78274d09c49d664d72a477af35a879ab1581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d62a78274d09c49d664d72a477af35a879ab1581"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
