)]}'
{"id":"openstack%2Fglance~873456","triplet_id":"openstack%2Fglance~stable%2Fussuri~Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f","project":"openstack/glance","branch":"stable/ussuri","hashtags":[],"change_id":"Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f","subject":"Update the RADOS authentication link","status":"ABANDONED","created":"2023-02-13 09:52:30.000000000","updated":"2023-02-13 14:13:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":873455,"cherry_pick_of_patch_set":1,"meta_rev_id":"67a09ba1761467ab479250472b6f83d39da558a8","_number":873456,"virtual_id_number":873456,"owner":{"_account_id":35025,"deleted":true},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-13 11:05:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 11:05:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14a8071de562a02788da859b54110fb9eeef91bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2023-02-13 09:52:30.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"fb91bb1312eaddf765ead5020b2d0c754ee24007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:05:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dd20c20d63f4bd293be5c50c758d7ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb8df6fce854182b000a1118914dd0e : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83cf9368e1f742468719af1369265773 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"67a09ba1761467ab479250472b6f83d39da558a8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-13 14:13:15.000000000","message":"Abandoned\n\nThis branch is too old, we will not be merging this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88f80d2fc0a57ec3c78220ba2602e120be60d54e","revisions":{"88f80d2fc0a57ec3c78220ba2602e120be60d54e":{"kind":"REWORK","_number":1,"created":"2023-02-13 09:52:30.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/56/873456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/873456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/873456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/873456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/873456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/873456/1"}}},"commit":{"parents":[{"commit":"7ffbbebd8815893b47958908be79cddc1a677ba0","subject":"[stable-only] Remove glance-code-constants-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ffbbebd8815893b47958908be79cddc1a677ba0"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-11-22 05:46:03.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-02-13 09:52:30.000000000","tz":0},"subject":"Update the RADOS authentication link","message":"Update the RADOS authentication link\n\nThe original link has failed, update the link\nhttps://docs.ceph.com/en/pacific/rados/configuration/auth-config-ref/.\n\nChange-Id: Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f\n(cherry picked from commit 3f9d58dedd5e5f5cd5034abbc86778dcf5d4c883)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88f80d2fc0a57ec3c78220ba2602e120be60d54e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88f80d2fc0a57ec3c78220ba2602e120be60d54e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
