)]}'
{"id":"openstack%2Fglance~873458","triplet_id":"openstack%2Fglance~stable%2Fstein~Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f","project":"openstack/glance","branch":"stable/stein","hashtags":[],"change_id":"Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f","subject":"Update the RADOS authentication link","status":"ABANDONED","created":"2023-02-13 09:52:59.000000000","updated":"2023-02-15 05:50:09.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":873457,"cherry_pick_of_patch_set":1,"meta_rev_id":"2e8f64012e90386eebbe3ff061bcbebf3943cd19","_number":873458,"virtual_id_number":873458,"owner":{"_account_id":35025,"deleted":true},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-13 11:01:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 11:01:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 05:50:09.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"}],"messages":[{"id":"bd2b056c4fbcda46f588c2a112f8d6177b27a5d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2023-02-13 09:52:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"a6320d33e5b145278617609e3b1404c9ca4d4d98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:01:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fabf74ae61df447b92083bb5f25689b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4cb41c081644c7b4ff67e18435cd37 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c44b238170b4a1c8a5f8d3c07e08f9c : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"159d45be1d69405f0d3c11afff15ec9d958c271d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-13 14:13:22.000000000","message":"Abandoned\n\nThis branch is too old, we will not be merging this.","accounts_in_message":[],"_revision_number":1},{"id":"2e8f64012e90386eebbe3ff061bcbebf3943cd19","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-15 05:50:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4b1bc6fb87dfb7ec31079de2aee2e75ef1f88a9","revisions":{"e4b1bc6fb87dfb7ec31079de2aee2e75ef1f88a9":{"kind":"REWORK","_number":1,"created":"2023-02-13 09:52:59.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/58/873458/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/58/873458/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/873458/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/873458/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/873458/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/58/873458/1"}}},"commit":{"parents":[{"commit":"149ea050cc58f39eaf9b4660bb8f0271b99d03da","subject":"Staging area not cleared if image is deleted while importing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/149ea050cc58f39eaf9b4660bb8f0271b99d03da"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-11-22 05:46:03.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-02-13 09:52:59.000000000","tz":0},"subject":"Update the RADOS authentication link","message":"Update the RADOS authentication link\n\nThe original link has failed, update the link\nhttps://docs.ceph.com/en/pacific/rados/configuration/auth-config-ref/.\n\nChange-Id: Iecbcbce1cd8d02cf4db83f2178235fa14e2e1b4f\n(cherry picked from commit 3f9d58dedd5e5f5cd5034abbc86778dcf5d4c883)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e4b1bc6fb87dfb7ec31079de2aee2e75ef1f88a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e4b1bc6fb87dfb7ec31079de2aee2e75ef1f88a9"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
